)]}'
{"id":"openstack%2Frequirements~804609","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I8b5a71e6e10938599f5c1259d27fa8c5463ca614","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I8b5a71e6e10938599f5c1259d27fa8c5463ca614","subject":"update constraint for os-net-config to new release 14.2.0","status":"MERGED","created":"2021-08-16 08:03:47.000000000","updated":"2021-08-16 15:28:42.000000000","submitted":"2021-08-16 15:28:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804609-new-release","meta_rev_id":"220b1b95e17fedb4af0cc97cd2e3f839225af150","_number":804609,"virtual_id_number":804609,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-16 15:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-16 14:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-16 09:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-16 14:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 08:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-16 09:34:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-16 09:43:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-16 14:01:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4cc40c77ae288394a71ae2e7f1b996d5c63baec4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-16 08:03:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6d07d87785c720d789a48175203efc4901f8d59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 08:24:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3180a76cf69b4162a896394ebb505627 : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":1},{"id":"806ed3562f7477650ba88b84e93031a26d00d637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 09:34:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b952fc6b18ee4e28ba6beafcbb2a7b43 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c5505985cd54c1fa7da0c4d270bc64e : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bdc5cd31a564cee9b5a6449e2b140df : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87951a5c77934be888795d2d91d746c6 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/daf21d0a87264630b0300689df7fe61a : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ee55833021b430889f31129c37cbc69 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/709a7c3c778c4e76a45ed032dad1c78d : SUCCESS in 9m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7070a74daf96464b98ee1a4e7b3bcd48 : SUCCESS in 8m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f10b9488144d4ce18e843adb1523d10c : SUCCESS in 9m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1dcb3ab09ed043bf85f0020e3d6328cb : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/007c946378ef4726a719729aef9d9030 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7af739dde79487983d621c436434940 : SUCCESS in 9m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/576398f990794470a3cbb487178c8857 : SUCCESS in 8m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/44a3965cda4f48f39abfb469815c034f : SUCCESS in 16m 38s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/baef7395170b4b9982e66ecbfc9778fe : SUCCESS in 6m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/469f11262db042699588f6e5553d3577 : SUCCESS in 22m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f5dbbd0d3d943089d5fe34dab95651e : SUCCESS in 13m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2855ec1931e04151bac4bc72364929ff : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73554df700244b4d8a67f92302b43d6c : SUCCESS in 19m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/62223f3a2c764c128eeb4804665175a1 : SUCCESS in 7m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/167fa212bc674bee8b97b684cd73e061 : SUCCESS in 8m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0ce555f656ac4ecaab20395be8725afc : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92bfe91f80714ee8b2a4d9eceeced4b7 : SUCCESS in 53m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80b2a4bd49ef4f4b8a25401da07d8a43 : SUCCESS in 1h 28m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cae726d5b0ddf582de899c41e69adbda8a2979d9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-16 09:43:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"929e6ade516e6cf0e9a9177d86ff95d395686c36","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-16 14:01:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ab1fe4759385f1ce99a2da84ea65e0ed84a408d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 14:02:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57ab9a82864acf376f7ed671b943d391ea5d11fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:28:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f7fe426d9b4e65bae1e0ad59468cdb : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a0ec17681b541b3a0e6304c0f479f0b : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0090930b92ba4f18b11ec97860534c1f : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b61eaa9170b4482b8fcea9e189aa753b : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/211dabfc56504c1ea01b2b9ed5082272 : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/05cdfc87e7af44a9814fb288b318c857 : SUCCESS in 12m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0acf24327f6b40718d4c0ada12b072aa : SUCCESS in 12m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9288584292b54fdea6003e3fdf08294e : SUCCESS in 10m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/96a9e458b0984e96a6d53a411670877b : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/60cd4e1685e547bca5bbab290f493afe : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/779406abd7fc47019b91c099bdde3214 : SUCCESS in 9m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e40e70195b284a87933eee767c81873c : SUCCESS in 9m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c00c21cb03de43f89328feb53f57e656 : SUCCESS in 19m 30s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/96fe6c36ae6749ca915f13e3e268e91e : SUCCESS in 6m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5ed6cda09e0f421ca92619b96f5ffbc8 : SUCCESS in 27m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c11e3af7e2444aad942c3f058a694173 : SUCCESS in 12m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0c8557018bd4df28acd39da195d8fdd : SUCCESS in 5m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7325926e66a74a4c83b4494880a95e36 : SUCCESS in 22m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/364fa35fc9bf4bf289a85cbd7411fa48 : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db0f8873dd3142429d1cfd1d11cf49d8 : SUCCESS in 8m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7345fcde916f4677ac3d2ba13f34b654 : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35cc45fe10014ae78b310935a2da4f48 : SUCCESS in 52m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ddfb65e70c843ffa98a3da94fa89dcc : SUCCESS in 1h 25m 27s","accounts_in_message":[],"_revision_number":1},{"id":"220b1b95e17fedb4af0cc97cd2e3f839225af150","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 15:28:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d1bf38f84f3d52d8372029f2051b0686176bf4d","revisions":{"1d1bf38f84f3d52d8372029f2051b0686176bf4d":{"kind":"REWORK","_number":1,"created":"2021-08-16 08:03:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/804609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/804609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/804609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/804609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/804609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/804609/1"}}},"commit":{"parents":[{"commit":"6a95003769be77dcb2d636d58684612e1824fd44","subject":"Merge \"update constraint for tripleo-common to new release 15.3.0\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a95003769be77dcb2d636d58684612e1824fd44"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-16 08:03:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-16 08:03:46.000000000","tz":0},"subject":"update constraint for os-net-config to new release 14.2.0","message":"update constraint for os-net-config to new release 14.2.0\n\nmeta: version: 14.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: I3a2b5e7296fefa26c100f7726ce364e7179e62e2\nmeta: release:Code-Review+1: likui \u003clikui@yovole.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I8b5a71e6e10938599f5c1259d27fa8c5463ca614\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d1bf38f84f3d52d8372029f2051b0686176bf4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d1bf38f84f3d52d8372029f2051b0686176bf4d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
