)]}'
{"id":"openstack%2Frequirements~804877","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ia0c060cfed34dc51af96d12f57e34f09e31a0c5d","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ia0c060cfed34dc51af96d12f57e34f09e31a0c5d","subject":"update constraint for oslo.db to new release 8.5.1","status":"MERGED","created":"2021-08-17 13:49:32.000000000","updated":"2021-08-17 19:29:52.000000000","submitted":"2021-08-17 19:29:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804877-new-release","meta_rev_id":"7053324b09c7dc22154bcae3e5131a769cc2541f","_number":804877,"virtual_id_number":804877,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 19:29:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-17 18:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-17 18:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 14:08:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-17 15:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-17 18:00:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"cd9ea4c80c26ce8d5f4986ba38acb971ca632804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-17 13:49:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5a46e2b3070082356d029dc061c4533ff09303a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 14:08:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/62265dee6ad94da39b513a332cb8d256 : FAILURE in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8de85fd974e83b92ec573a4b84f42f42eaf226fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 15:25:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e3e311901b40cead4839ad894e1272 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37a8b438c4a9476cbcfa941cdcc31e2d : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/340d614fad1545a6957be0d215ae09e8 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b04eb4f5a604080b9564c8d35d92c5a : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b61673eb792f46ad8491e8062d13b696 : SUCCESS in 6m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d178590f6004436bb12d7bd987fb990b : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b85c93fefa747bba3ca1bb7250fbf16 : SUCCESS in 10m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0b3bc77050b640f6be3ef6dcac965d6a : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd61313ff863455799a7e35baf437e83 : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/43e97d93eb5741c08a11b6c91eb27852 : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0641f879481a4d0891a023c7076bc2af : SUCCESS in 9m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/910661024a234671a4f06371d87fd3ae : SUCCESS in 10m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6a92830edef8459eb3f4523d8d7aa709 : SUCCESS in 9m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/80d0f9de7ef24a078346c98f7dfe1f55 : SUCCESS in 12m 52s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/36c83a51079f4971a974ec9518b718ea : SUCCESS in 5m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/003fc8f842cd45c98b2738cbe709f111 : SUCCESS in 33m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e2f8ecfd5f8c46c1a5597f58ebe5a458 : SUCCESS in 12m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c8c66819732546df8263f4161d435c7b : SUCCESS in 6m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/776618d503ce4f44b07e034b7cfc8927 : SUCCESS in 17m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e76306075bf54f999a1a8d1261cb0213 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55662bb8973b485a97ba3e428f10ccb8 : SUCCESS in 6m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/84707502dd97477788674dbb5c51de16 : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c91dfb410e7840d48454135aea346bee : SUCCESS in 43m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9219289a1ed4eebaa2e681bb44d9213 : SUCCESS in 1h 34m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9087a90d644ae54d6be6043ce4e60fb1d6f8467d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-17 18:00:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67dd2c52d6a871df3eb2110350076a0721b46e81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 18:01:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62bf542b8bcde5160b87689938cf16707543e25a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 19:29:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e16c9efb7d4c26818ade1a16ac8923 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f66860c7b56644baa3499e63bd4cf7c7 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7dff12358ab437dbcb80dbc34e41ed6 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/365d7127b3f947019deeb95c0333d677 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c648a3caa37485c90b723aa4fea2eb6 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e4c28958e207451188f524dcfa9d84de : SUCCESS in 9m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b0ae1d0599d24c74ac751f92619b57f9 : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3a8e5634b3db4f999db084b7d346e586 : SUCCESS in 10m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/742d16c381144b189e9ffae659b2c0d9 : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7d08e45fbb26424583048d087b7b0eda : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/218980bda4cf4bea9537434eb57712f2 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/640d148597bb46c6b0849bfee45e4f3c : SUCCESS in 8m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da1472a684ee4f8ea6aa8d83ba578b1f : SUCCESS in 14m 20s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/416b1b6ec8614c5db8e57df21364372d : SUCCESS in 5m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/48ca40a66c884ba7baec50f5c6d10441 : SUCCESS in 29m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1c1ce0a29fde4cb6b3c52123aaec1f98 : SUCCESS in 15m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/38113f99ffce4cc38421bc7fd49b55f9 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4845ede9ff8340319787ca29a8391d0c : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4a1cccd30af04c3f9079ba99dc66c712 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8fac4f9049d4edd909fe6de547c0c4b : SUCCESS in 7m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/836601b1548f4ac0ad4b04d77adca9d5 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2288a5cca0a2419fb560a72270b53884 : SUCCESS in 56m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52c83293207e4b1f8d538f75a4e0469f : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7053324b09c7dc22154bcae3e5131a769cc2541f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 19:29:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3018b1f729be36cfb96d5cd71c9505bf32100e9","revisions":{"c3018b1f729be36cfb96d5cd71c9505bf32100e9":{"kind":"REWORK","_number":1,"created":"2021-08-17 13:49:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/804877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/804877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/804877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/804877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/804877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/804877/1"}}},"commit":{"parents":[{"commit":"1d1bf38f84f3d52d8372029f2051b0686176bf4d","subject":"update constraint for os-net-config to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d1bf38f84f3d52d8372029f2051b0686176bf4d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-17 13:49:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-17 13:49:28.000000000","tz":0},"subject":"update constraint for oslo.db to new release 8.5.1","message":"update constraint for oslo.db to new release 8.5.1\n\nmeta: version: 8.5.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: I333a89a2990861b75ee28539839c7134a3f809ab\nmeta: release:Code-Review+1: Brin Zhang \u003czhangbailin@inspur.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia0c060cfed34dc51af96d12f57e34f09e31a0c5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3018b1f729be36cfb96d5cd71c9505bf32100e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3018b1f729be36cfb96d5cd71c9505bf32100e9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
