)]}'
{"id":"openstack%2Frequirements~804965","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I53ddd8e3a4c943577e183527348ffaaf123a4027","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I53ddd8e3a4c943577e183527348ffaaf123a4027","subject":"update constraint for oslo.db to new release 8.4.1","status":"MERGED","created":"2021-08-18 08:32:19.000000000","updated":"2021-08-19 01:09:04.000000000","submitted":"2021-08-19 01:09:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804965-new-release","meta_rev_id":"4c1279999bb849da38187eb62b4b865d301b8795","_number":804965,"virtual_id_number":804965,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 01:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-18 14:53:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-18 08:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-18 14:53:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 08:51:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-18 08:52:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-18 10:04:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 14:53:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5923ac9996b0b5181a47e08a0568ee5f8e44779f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-18 08:32:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"291f3f2a13529d8fccdc45fec0ee3d8d0c780647","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-18 08:51:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"05c846f921eef4ff4e275bdb5b6931231143b3a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 08:52:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6d6619fa87d64f4ea3177650a4f3d832 : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b40ffd68d48a2b436ee4be0e4ceb9f180d4559da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:04:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda2118540de4ea5a94ddb95a177a2fd : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5829f44c16da45cbb8f0f962148be081 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3505fcef2cf249118afd3b7bb8bb8270 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58fa918826674f1da1625568aa73ede9 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77f6c12805424010bcadd7048548002e : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac907bab786f44f9878e7cbfe787d97c : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8b41bc768a5e43838cf7ce62b7aeec98 : SUCCESS in 10m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c98c5a1480a64e70afe95d52eaa1b7e6 : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/182eb7ffd08f42bca5b6721e15216f24 : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9edb53b3113402880ce04157d5d3d69 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f33d2a4db70a427c9a5a691eebe4d5e0 : SUCCESS in 10m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/61b9a320fdaa4b85b2f12c7e1937205e : SUCCESS in 7m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af2baf6d85a0413182e1f80ec7a9c5c3 : SUCCESS in 14m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/21727bee2b684c738c19bfb0a86c9490 : SUCCESS in 23m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3d712c6c5e144007805da7a939327df5 : SUCCESS in 13m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bbd1bddfd1da49408584843893977133 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc25f6241028408f90c0f89bf99aaf76 : SUCCESS in 17m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ba7d05e10360498f86c2083bf7545f06 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7bdd47280eaa49a285be8e2c04de6bee : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c59f214ae1f24600bf362838f87cb8a3 : SUCCESS in 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/891d1f63e1694643b6d35a2c1c94e090 : SUCCESS in 53m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71f4fea407a1494b9994cc0e34fb1e57 : SUCCESS in 1h 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"52c3d920cd47dbd6c1bef5863b8affb21e1aebe9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-18 14:53:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cfe054cf58111b711241a6c45de23f2553895d96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 14:55:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"945121b08b39d8ccd693c2e3d749a4373b92b58d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 17:30:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b3715feb4c34beab508df8483173c58 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b4023bd0be8403d802068d8b39fb0a3 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78cbefc9c37a479b8f683fcc0cd22a26 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a3e6d2fdcc044a589ad348c978daf65 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1834801f7bd84535bc250c61f989f9d9 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/754c641b2143429190d1ecbe6c1a9b13 : SUCCESS in 10m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9107a2f070af4381bca1ab08533183f7 : SUCCESS in 14m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/339c3b706b814b4ca2354b24eed63d29 : FAILURE in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2af5659eac0b4bda82ca6c5b2682d051 : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/08b59f46257c4d76aca93615ecb4c996 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eabd9a63f2ed49a68f44899dfb5414a6 : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bae9229d5ec2429b8eac555ec39dc70b : SUCCESS in 10m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96b4fbc2b658455bb14f01e147ebafbb : SUCCESS in 17m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7d340af5385c43a08c71f73168c6b77d : SUCCESS in 22m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c0c5f965d754da1b17d1303a8da5a5e : SUCCESS in 13m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba74139ce6204bcb83dc8a2b013290c8 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bfd134c93cf242949faab1a77800728d : SUCCESS in 19m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7c803b2e9e4a4e4f82a8727a80e9c68c : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7320d4ed2f584680b1fce17da9a0c995 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9a1c5d602d1c4ea2a9b18d40b003cc1c : SUCCESS in 13m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/486b81f690ec4be79797422d626cbf98 : SUCCESS in 57m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35446d5703a849b0968a16af0eb67d16 : SUCCESS in 1h 22m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a270dc6ca7d827fa4f6b476339b3d8000a7cdc1a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-18 17:36:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c5cbd22e008aa870d1a88de2cbc6c46480e1d0cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 17:56:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0b0be560290d4a64b1b17d4e167a0fac : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3179b8e44805ee49c472455127d2e6fb720041bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 19:20:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c3ee4e70ba45aa8fe4163d361388c7 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a3044d506684e5787597f85a4073fdc : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f33bf0dd707459f99016d5d45fcf689 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c45ed5b252149f8bd11e60d027c8303 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae1db8ea84044bfe8a0493ca77bd561d : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e626ffa65be54aea82dbaef254139058 : SUCCESS in 9m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ea117a6695c347a2874902f13220ed7d : SUCCESS in 8m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4ff9cca9127e4f85b7e1661ce4b0b25f : SUCCESS in 9m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bc5481cb1cd54268b42c8f9e203cb07a : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c9b688808069470da9265c681d7665de : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed01c6519f3d4626a31fc71ed79821e0 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2c57720c4dec4b68acce744933097926 : SUCCESS in 7m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7337124f6ed1464dbdfb95b5035eb845 : SUCCESS in 12m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f9114ec65aaf4d83a71729497a636db6 : SUCCESS in 25m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/af1a534cf3c243ab9af6057b7e6694ff : SUCCESS in 11m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a2b70bc6364143c0861adef3eaa87058 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b50318d1f50b45ebbc1e2ba45fe0f831 : SUCCESS in 17m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/867c40ccde794a3385b896050952748a : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b565fe1a6698461bb2e091a52127d21a : SUCCESS in 6m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bb2c1fd1b0384ec5b20b9cdce1c0868d : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/944aaaf010654c5f8c6b948b8d5f52a0 : SUCCESS in 52m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77c1999a2bd844008bc6722a371f9d28 : SUCCESS in 1h 42m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0d192836bef7bb670e89adc737c7c1febffa8635","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 19:21:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcc34e8d3d9bcf7d7330ffe3b43b207fd7848c5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 23:35:14.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4da09a5c1510d30c6d4d144ab821152ea02a935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 01:09:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a02c36cc8b94ded81ef93a076348964 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d943fcc373343beaccdcecdfe4acdfb : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53dba37a01d24de2b8dda3542d2f21c2 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc65d69d98854512aca29fc060938f3a : SUCCESS in 5m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8b68cc624d74b5d9e0c8fd9a575c549 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/165b54f9567646c2af0d9983ff0b9736 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9effca601afd41b8a0718c0adbd32866 : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fdda2c8bf85a4be1819f3855911fc43a : SUCCESS in 11m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/861ba44089f84dbdac7f186931a5179d : SUCCESS in 6m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/525118fe65bb49c3aaf36b5f22382848 : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2fb7215cf8a74475808cb74e5cdc1748 : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/24188ae01bca4bc2a98f90494d3e310c : SUCCESS in 8m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/531d9f5a5a154c229a1f9dee9f994c04 : SUCCESS in 13m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e46885b148ab40379538edacd1ddaf9c : SUCCESS in 23m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/46b83352ab7c40b6a875733f5a00f305 : SUCCESS in 12m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd268bfc2c4b4c859b0b9e57cdc6a642 : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0126fef640714100aec69000a25f1399 : SUCCESS in 19m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1c7bfbe2c06b4c32ae296741013fb2ae : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/610b9c539e0b48f8af917fd84086ec18 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c7a0cbed09ce447d9a163503d786296b : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2bbb2ed5b8b4302ba4523d5f97a4d91 : SUCCESS in 51m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/baac07bddc1d4f0d808e95e259f19abf : SUCCESS in 1h 29m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4c1279999bb849da38187eb62b4b865d301b8795","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 01:09:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"333de443832823c9ea65f6713cfcb70357f9cd96","revisions":{"333de443832823c9ea65f6713cfcb70357f9cd96":{"kind":"REWORK","_number":1,"created":"2021-08-18 08:32:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/804965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/804965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/804965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/804965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/804965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/804965/1"}}},"commit":{"parents":[{"commit":"1402df485a5f5f47d245da1d31e1879f32346b2c","subject":"Merge \"update constraint for os-net-config to new release 13.2.1\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1402df485a5f5f47d245da1d31e1879f32346b2c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-18 08:32:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-18 08:32:18.000000000","tz":0},"subject":"update constraint for oslo.db to new release 8.4.1","message":"update constraint for oslo.db to new release 8.4.1\n\nmeta: version: 8.4.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: zhangbailin \u003czhangbailin@inspur.com\u003e\nmeta: release:Commit: zhangbailin \u003czhangbailin@inspur.com\u003e\nmeta: release:Change-Id: I1abe1109d2cf2cd2658324b06c7d6384c6f0d60d\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I53ddd8e3a4c943577e183527348ffaaf123a4027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/333de443832823c9ea65f6713cfcb70357f9cd96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/333de443832823c9ea65f6713cfcb70357f9cd96"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
