)]}'
{"id":"openstack%2Frequirements~805199","triplet_id":"openstack%2Frequirements~master~I1acd4ff7710ad5c8028f7a0470db7634b5ee8a9d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1acd4ff7710ad5c8028f7a0470db7634b5ee8a9d","subject":"update constraint for automaton to new release 2.4.0","status":"MERGED","created":"2021-08-19 12:44:23.000000000","updated":"2021-08-19 18:39:17.000000000","submitted":"2021-08-19 18:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805199-new-release","meta_rev_id":"f741b9d5f8dce7b8ae56ff19d358b63b8af9eea8","_number":805199,"virtual_id_number":805199,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 18:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-19 15:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-19 13:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-19 15:19:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 13:01:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 13:37:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-19 14:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 15:19:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"67335cee278e0928bd5b524b54d16e39c7db1813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-19 12:44:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98bc86d183a9c78ef0fa692b17ab73de37be02e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:01:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bc2f379824634c1498f0e0b7eb381fe5 : FAILURE in 14m 53s","accounts_in_message":[],"_revision_number":1},{"id":"47f3cfd85c7e82cef74589e1d0d80ac47a1da5d6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 13:37:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61f9f366ec0681c4e46c9f3edd3a1c97df7e3db8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:04:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ba3461c6d9e4cf48fc36adf169a0199 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd3b1e32abfc49cba9c96fe8deeb2efc : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa14e7a70bbe48df9370ae99e69e4e85 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fbb0e5a998c469eb00f2852bab7accd : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83b7aed912a046cf9f8d09e1894dbc53 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c03f270d580e459598e77c2b98c635d2 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c916fba3c634727b4f798cb747bc49f : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/291eb9c3ccc7454f8a8223483886406e : SUCCESS in 11m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/052bd7bce8a4463c9f82c374718dd527 : SUCCESS in 10m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/04ee1f7e20a848be8e77dba279bac325 : SUCCESS in 6m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f0f2547314cc4526956500a8bf0d0e2f : SUCCESS in 10m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2bbf1fbeb1a34bfbb5b89627cc6a6ae1 : SUCCESS in 7m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7f834293f8d747bbbf86db27c23d71bc : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c8e95d1e82c4400c8440c316ca237a3a : SUCCESS in 9m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b2e2117794594f2887cd25589abd84ef : SUCCESS in 9m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e8dcd2e3f19544ff8ca80842a26a7a0e : SUCCESS in 16m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9622184fb8804566806fc7f98d5c439a : SUCCESS in 11m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eb8161dea76c4f9c92e6687741f272a5 : SUCCESS in 6m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/88a8aa30f0da4a52b609dc7497f5dded : SUCCESS in 26m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/230aa2c712ad42ccba5c42e438c2d2a2 : SUCCESS in 14m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89bcfba648f74a4980cc29e1a99064bc : SUCCESS in 5m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e0eb3ce16b4949d2a0f7b1e089d74da7 : SUCCESS in 19m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d391146bf8f9487da4134446a08fac44 : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ade0a19f67a546b38cdff2b767f57e53 : SUCCESS in 7m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4f911f50782b4385b5796d87d75e2636 : SUCCESS in 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9c87552056c449fad8b78bdde0bdcbf : SUCCESS in 50m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19d48fc962614903878dd5fe2bc8dcb5 : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9cf97a01c99f4176c1a9bd16c48b15e52d973786","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-19 15:19:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66f3ed66ecbdc345b5ea1c626125dd3ea5c7994b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:21:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aec762569ab21c4fc7612643e75815d43637958b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:39:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40de082e8546405c93d82ed3ebf6ecb2 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834a92e2891f434a9524b99321761060 : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c8a743285894974befb9e104be2d053 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2adb9f0c0ec44a629d52f09435d49702 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/38080f2419e040e994480830e2ecbce9 : SUCCESS in 6m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29f4717c64f04e90bf48fba5535f73a8 : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8dd3ba59717f44d9a0bf8e423eaf86cc : SUCCESS in 12m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78b91269d9e148b483e10037e9b1461c : SUCCESS in 11m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7000ced82bd94442a6327c88b41d7fde : SUCCESS in 10m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9d16e096594d4f559f37f256de076828 : SUCCESS in 10m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/18b9dcd04c0a493c82eb15135a421dec : SUCCESS in 13m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/306b8eff9bae47cda88ab33ec56254da : SUCCESS in 11m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7055575b88c14e53b9be8529a75f2d48 : SUCCESS in 11m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/25b6d5168e0b428aa56c70ddcff2f1d7 : SUCCESS in 15m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/62ea5937818146b1adfde0935b3ecbff : SUCCESS in 11m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f67dff7e8b884c09aa09386a7c393e11 : SUCCESS in 9m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/49c9939c5520498fa9270c6bd5a732fb : SUCCESS in 26m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/edd7df6859214987af7c3354dc467389 : SUCCESS in 16m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/723b72929bc04eabacf303548c41ceee : SUCCESS in 5m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/79e37637829646a8a2dc1dccec2c7fba : SUCCESS in 21m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/75fbffcc0aac4096abca9a6f99bccd51 : SUCCESS in 7m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7b4f2dfd3a77477288166a419c72a605 : SUCCESS in 8m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5d16e31c67244fbb912d2b5c619722d1 : SUCCESS in 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90161fabaeb14af1aadd5f11bd3b4485 : SUCCESS in 46m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/904b0837e1c74b1f8e25f7af86ccefc1 : SUCCESS in 1h 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f741b9d5f8dce7b8ae56ff19d358b63b8af9eea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:39:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d28ba054aa438d830ba66d9c8a13f6110ae31327","revisions":{"d28ba054aa438d830ba66d9c8a13f6110ae31327":{"kind":"REWORK","_number":1,"created":"2021-08-19 12:44:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/805199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/805199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/805199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/805199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/805199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/805199/1"}}},"commit":{"parents":[{"commit":"ad5b9d5625ae3108d04f9cbeafc75363cbbd039c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad5b9d5625ae3108d04f9cbeafc75363cbbd039c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:44:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:44:21.000000000","tz":0},"subject":"update constraint for automaton to new release 2.4.0","message":"update constraint for automaton to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I83550e91db0b8edb77cf03d0c178d7934a1ce634\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1acd4ff7710ad5c8028f7a0470db7634b5ee8a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d28ba054aa438d830ba66d9c8a13f6110ae31327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d28ba054aa438d830ba66d9c8a13f6110ae31327"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
