)]}'
{"id":"openstack%2Frequirements~805201","triplet_id":"openstack%2Frequirements~master~Ic9986138fd055d32913027cbd5313cd4f7e9f22a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic9986138fd055d32913027cbd5313cd4f7e9f22a","subject":"update constraint for oslo.versionedobjects to new release 2.5.0","status":"MERGED","created":"2021-08-19 12:46:57.000000000","updated":"2021-08-19 18:39:19.000000000","submitted":"2021-08-19 18:39:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805201-new-release","meta_rev_id":"930dfcb2ae7f9ca3957d888a63c78108f44775da","_number":805201,"virtual_id_number":805201,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 18:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-19 15:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-19 13:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-19 15:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 13:03:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 13:36:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-19 14:02:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 15:19:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e59c21b3a91158005a0379d350b553386f6bbc7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-19 12:46:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5078f8e9438432adbafad76ee3ef85be6e448825","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:03:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a5fb59ad25ec4cd9904d53de55bc6c1f : FAILURE in 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ca2bd54dcf83c4644d75b5db56f428046e87a79b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 13:36:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9984c0e3bfec778c8c1a2f0930e8c7cbf8291f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:02:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6f151eba154d96992a22274b56d48e : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ef7dbc836f842eda20ecaf9ad13e28e : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5d2f0e3278c47539d2e22358961d38d : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fd60fb1883e433483d0483e22deb144 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0214c69b1ba44cd783a8bd389fffd826 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4bfde62cf4164948a948d20cdd5ddab2 : SUCCESS in 6m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3c9b3ca7ea764dc98ff12901c9deec70 : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b7cf4679638442e59236f497826c3861 : SUCCESS in 13m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e93f9be933948cd90f4c8b10af58029 : SUCCESS in 10m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/80550817e45b44c0b08a3cce5ed8b5bc : SUCCESS in 8m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fa43f75f52fa43898ca49e6528a92a8b : SUCCESS in 9m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/95f3772388cb4fea97170aabdc5c537f : SUCCESS in 7m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/16dc34e5be3543c78494887f3433ed08 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d264832c58f54418b0968e502d0bb091 : SUCCESS in 10m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7427d7973be142ae95a64f0ab1870b3d : SUCCESS in 10m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/17b33a7991e34cad8d03d5fd06e87847 : SUCCESS in 18m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6e0ad27d253a43de8b55b9da2339f5fe : SUCCESS in 10m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e64864bed5fb4dbd8f85250323c77b1c : SUCCESS in 6m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c51bd60930e64d8a8b17f6b2741be0b7 : SUCCESS in 24m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bdd16418bbed4b56a07558bb18d8ceda : SUCCESS in 13m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aef9eaf7db5d4233ab097f307de2c265 : SUCCESS in 5m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0c8860a09e5d4384b11333fe9a06538b : SUCCESS in 16m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d5b40054329a426f8799d64b6e2e9041 : SUCCESS in 7m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b746d6ae17ce432a86c4450e3667b282 : SUCCESS in 8m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c9429c2162ad45a2a06d9b0303b30a67 : SUCCESS in 12m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/277c54778e9c4786b297bc1549fa85b3 : SUCCESS in 45m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f37218a91b164284b6c3e95fe519efb2 : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"6b322b9aa2e7d0b77375ad8e6eaffe18a9824e8d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-19 15:19:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3653673d8230ac35fda0f31fa261deca22242358","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:21:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b51bbd108a5bd707f51576af42b5e01cddb21e5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:39:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e74694bbf9464f9f4e4da36be2cb76 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/826fc504fc3e4aaf86b50c0039baa168 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f512312a4234bd0b410d0d6483edf75 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b3aa89796e34e65834cb2381c9ed695 : SUCCESS in 6m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9636527231c4ef6bec1371eef238a39 : SUCCESS in 6m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/666c61ce613b491fb2ec80217e7af6d0 : SUCCESS in 5m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9948ce25263b4c008558edde22250a98 : SUCCESS in 11m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aefcf2c06c324f1bacb6d31aae97d1fb : SUCCESS in 10m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eea95a3c7004430991b531ae31482fa2 : SUCCESS in 10m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a6f9792cbddd4ff6b284b6290558f09c : SUCCESS in 7m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/74f8e79978b640f6af1989a18216b5c8 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae23a5462151416481671c251dc2f67d : SUCCESS in 12m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/01b76055af8847b18fac610f1f846c55 : SUCCESS in 9m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cb146df56c6040faa8e79145ab2aaeea : SUCCESS in 16m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/21711b7cfdc14293abb79d1c599f4e4c : SUCCESS in 11m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/09aabae51d04462282b8972bb2b54954 : SUCCESS in 8m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2ef2cf02c8e24c36af732f1e7d562a94 : SUCCESS in 30m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fbadbe7547d042e98e248fb8295aba69 : SUCCESS in 15m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d676e1a13c8746aea89160ad3f14e0a5 : SUCCESS in 6m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/591243cc55944ba28326968dfe7bb153 : SUCCESS in 18m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/615bc231867440b19af282f5aa60add4 : SUCCESS in 8m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c945ccbdc0ec476e971bf61e3a0e62b6 : SUCCESS in 8m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/88cd58094cd74f0db882b61301c10858 : SUCCESS in 15m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5325c003b15f44fd9d89c60abb81c41d : SUCCESS in 47m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b89cfd2823fe4100a775b561f4fb4f70 : SUCCESS in 1h 18m 46s","accounts_in_message":[],"_revision_number":1},{"id":"930dfcb2ae7f9ca3957d888a63c78108f44775da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:39:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37e22f48ac685056832ba3037482856120e88075","revisions":{"37e22f48ac685056832ba3037482856120e88075":{"kind":"REWORK","_number":1,"created":"2021-08-19 12:46:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/805201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/805201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/805201/1"}}},"commit":{"parents":[{"commit":"ad5b9d5625ae3108d04f9cbeafc75363cbbd039c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad5b9d5625ae3108d04f9cbeafc75363cbbd039c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:46:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:46:55.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 2.5.0","message":"update constraint for oslo.versionedobjects to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I27a4ec0615c6ca7291b25ad0aa9663503404c4c0\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic9986138fd055d32913027cbd5313cd4f7e9f22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37e22f48ac685056832ba3037482856120e88075"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37e22f48ac685056832ba3037482856120e88075"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
