)]}'
{"id":"openstack%2Frequirements~805203","triplet_id":"openstack%2Frequirements~master~Icb2c9d2972fc2356772a16edc80e3ab0582f4a6c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icb2c9d2972fc2356772a16edc80e3ab0582f4a6c","subject":"update constraint for oslo.utils to new release 4.10.0","status":"MERGED","created":"2021-08-19 12:48:35.000000000","updated":"2021-08-19 18:39:28.000000000","submitted":"2021-08-19 18:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805203-new-release","meta_rev_id":"15ee16369664f7fa611a37e3db4845b1f95209e8","_number":805203,"virtual_id_number":805203,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 18:39:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-19 15:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-19 13:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-19 15:19:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 13:03:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 13:36:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-19 14:17:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 15:19:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e9bf0d1114083130acaff0cbddf8405bea4510ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-19 12:48:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c67ed9ba3d9bf2296de1f5e9bb52a9e28cc5de3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:03:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/20a6c83954ee49e4929dc774ccffba02 : FAILURE in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f2f451f3bc6924bfefb53c7321fd00de48f7a76b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 13:36:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0e0ae8278bbef1b20a6bdf23e17a8411330a497","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:17:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e5d591e36cd418993dbebbf0c611c95 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9faf1b921a40629188ef9bc439f9b4 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9c32ba20ce8470b9a48dba0d29b1b96 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e70248f878d4f61bed742fed4e07a3e : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bb70c66608f47229e964e3a8ddd6fbd : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0238a4994b2b4717bb14f5c10467dac3 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1418c96485ac4614b65b02451745adb9 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6f136f1b10e4244a0529bfb00a1e8fa : SUCCESS in 12m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42974e3b8ed44e548e5872d22b91fbf5 : SUCCESS in 10m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/89128ef4186b4d1fa9452452b5db5256 : SUCCESS in 8m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/25567272580143b584545fbc207492c7 : SUCCESS in 9m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/865fdba50b384039b28861679803692e : SUCCESS in 7m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7075ebfb1aaf44389bf23e9e6442dad8 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62b32782042b44108803208060109b81 : SUCCESS in 11m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/689897957e694d5d8744f254fbd7d2c6 : SUCCESS in 10m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f85ee50f4c434f098daef83d565b1813 : SUCCESS in 14m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f588ec80fd1b479098c164e7c5661c87 : SUCCESS in 10m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8422ba297d7f4f9896cd63b490375b20 : SUCCESS in 6m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5fea86c2248241c9a6e58efbb748a1ab : SUCCESS in 25m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c215ff5185cc475e820c47bceaf0bbdc : SUCCESS in 14m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c3d4d50571c48fd8dec5d17deacdcd6 : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1656cddd9def4c6a81bf3dd5e87327f1 : SUCCESS in 17m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8658158e969c41009aac37795612e85f : SUCCESS in 8m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/812823bd5db34b4db731d6817bd68113 : SUCCESS in 9m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6cfcdb83036247ba98b3c968bea90069 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d1ef3aed6a14a8cb8803ddf36eb0bf5 : SUCCESS in 47m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c349ff4adf374f1dac40a8b5fe785caa : SUCCESS in 1h 24m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a568ac468dadb6d9cd56ade51bddce479ea07ab9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-19 15:19:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7379d08ab7ad85403aeb39afb989d3c22c0c3be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:21:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56cdc0ce9cb403c0aa21dfc24988536cc4fc194b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:39:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b03bd7f35746a79fa8e287bce78a1a : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9103a6d8d0d849aab63030a02382b3ba : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/417f4e575d9449b7b8c7c7fb361b2c10 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d0bfadf383d455d8e06532b98ff7be8 : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed21c5dd11e3434eb58b0aade73d584b : SUCCESS in 7m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32b158d4fbec44f6b57e0e311bc38d9d : SUCCESS in 7m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af87e72e2fd14871a775e399bfb7e030 : SUCCESS in 12m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35717021a82a4c4c944e618db8829850 : SUCCESS in 11m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8000d40a20de4d45bffb939ca623d688 : SUCCESS in 11m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a3c41b1e29aa4bf7910212a1a7b5b7ac : SUCCESS in 8m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a495106a32a4d83a1668b49f6a8add5 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd7a357d75334b1c96706fdd26c1f564 : SUCCESS in 10m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1242d27bce3748aa8d2778e74f287158 : SUCCESS in 11m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/174f88a7e40a4e9ca7e0e0bcb8976992 : SUCCESS in 16m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9f94adbd7f30475e94d29d44b4ff2adf : SUCCESS in 12m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bf088a95d8b041b5ab35f3316fe0d07d : SUCCESS in 8m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d864b8670af04aaf8d2ad5bde1fc5e6b : SUCCESS in 30m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5d70d90286d146538f4dda0661e7ddd1 : SUCCESS in 14m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/503b512bdcf948eb9eea134b3f701818 : SUCCESS in 8m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ddcf5e4c114642a985eff17547cd59a0 : SUCCESS in 20m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bb9ec80b9ca44a5db256ea1d5c3dffcb : SUCCESS in 8m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/925c7f4a67254b8d8936f817dfc6815f : SUCCESS in 9m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b7258f4676474e3489808a97ab629fb8 : SUCCESS in 14m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efdac5493e97490d967dba94bab95262 : SUCCESS in 47m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbe9514daa304dee9a2dd67faffcef41 : SUCCESS in 1h 21m 45s","accounts_in_message":[],"_revision_number":1},{"id":"15ee16369664f7fa611a37e3db4845b1f95209e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:39:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"557e9ea48b4beafea97bd66f84e239234b4706c5","revisions":{"557e9ea48b4beafea97bd66f84e239234b4706c5":{"kind":"REWORK","_number":1,"created":"2021-08-19 12:48:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/805203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/805203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/805203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/805203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/805203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/805203/1"}}},"commit":{"parents":[{"commit":"ad5b9d5625ae3108d04f9cbeafc75363cbbd039c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad5b9d5625ae3108d04f9cbeafc75363cbbd039c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:48:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:48:34.000000000","tz":0},"subject":"update constraint for oslo.utils to new release 4.10.0","message":"update constraint for oslo.utils to new release 4.10.0\n\nmeta: version: 4.10.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ife237e0d09beb2040af5934965100123258db223\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Icb2c9d2972fc2356772a16edc80e3ab0582f4a6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/557e9ea48b4beafea97bd66f84e239234b4706c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/557e9ea48b4beafea97bd66f84e239234b4706c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
