)]}'
{"id":"openstack%2Frequirements~805205","triplet_id":"openstack%2Frequirements~master~I2b9b9edf5384ae66ed1e81c817d3e7b12841b00b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2b9b9edf5384ae66ed1e81c817d3e7b12841b00b","subject":"update constraint for oslo.privsep to new release 2.6.1","status":"MERGED","created":"2021-08-19 12:49:07.000000000","updated":"2021-08-19 18:39:22.000000000","submitted":"2021-08-19 18:39:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805205-new-release","meta_rev_id":"29135a31bb2f96222a01db7f889b0fa47c5547aa","_number":805205,"virtual_id_number":805205,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 18:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-19 15:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-19 13:36:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-19 15:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 13:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 13:36:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-19 14:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 15:19:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c90c8b2d7e23b061a4e75a7011c5a6724061769d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-19 12:49:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af9df314611b746ce36eae6bade253f32fcd1069","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:04:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0e6af4d02e8845f19f1b90e808b26e55 : FAILURE in 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"dd4ae3c663f349d6c48443c3409f5db453daeb72","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 13:36:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8518ede56b19823e03ae10f5dbbe9d7777e07062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:13:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1376c6fd24f94fcd9ee1bbd8cbb64e11 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e90e2b6e37e4e37aa543f62300161c6 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/604306c1f40a48aa988fbfbbfc302262 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be053daf197347bd9309d1b1e01d9733 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7fd847fb6d64f36b5fb601e4ef7931c : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c79bc834613d49a2adf6351c8547581a : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97b135028faa4c3eb69a7f8565641a91 : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4bf6769d340443afb3290fb0941caf46 : SUCCESS in 12m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bcdf8ccb5f5e4f1d9a9e005bc36d7442 : SUCCESS in 11m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2b6240a3f7b2445a949b71fd5be509fa : SUCCESS in 8m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/caa984edbab54127a9170ae80fcf28e2 : SUCCESS in 12m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e8b341c785004c42a269ac00c7699bb6 : SUCCESS in 7m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/353e2fe2d038498385a6e8965480adca : SUCCESS in 13m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d6bfa3454e94809b025eb572c1a79d7 : SUCCESS in 10m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/237f58a69ffd4fc98a6ea6402028564c : SUCCESS in 11m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e39f8b040dc8466cbdc62b873a648efe : SUCCESS in 14m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a1a21afc62a54bd49bd816561c93a50b : SUCCESS in 10m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0c1e6fbde62d4a998b4ca5e1ba40e9a3 : SUCCESS in 6m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9c874939540e492c8f59b3b92d2c1939 : SUCCESS in 23m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d40684441ea3440dbb508051204fb92c : SUCCESS in 14m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7b285ba775f244cf83291a1e95c8b278 : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/740cadeadb7840839431f958b79a7c71 : SUCCESS in 19m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/756238ed8cd64e77b9b41344b92435fa : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b28059bd376499b874e7649859d79e9 : SUCCESS in 9m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/62f91ed5f32d4d6e9f207f21ff9a544b : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/417a938fc75c42a7912144a6b419a05c : SUCCESS in 47m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0c3c249f5a04fd5bcbc3b37938a527f : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f788566c20c25a3950b0d0f247b146b364a62009","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-19 15:19:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcf016c5b61f99a909870e11fac3a67fb3873e2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:21:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7be5af2a1e507e30529c551acfa3474bf8b7a4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:39:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b59ff99ccfcd414cb6c5dc6539ba0d94 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fe4814f9c6c4cd0b733cb555a6cf347 : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f2cfb2c60ca4d5e836d67caa8b85177 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd3e673bdc564386ae6723ab39fca871 : SUCCESS in 7m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/251d839e23ec4dc9ba5cb2d77e00217f : SUCCESS in 6m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5538693d18014bdcb85c5a2d626d9c1c : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/478f95c6b87447b78c2f735a95c774ea : SUCCESS in 12m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d720a46ff69b457eb23e6883056dee15 : SUCCESS in 11m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ff0bceb277bc4493a8faada362e933f7 : SUCCESS in 10m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/641d97d0844b4c71a488b17e191d8be3 : SUCCESS in 10m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4f3c746d990f4b0f996559cb8f18bc4d : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c419abcda7d149a1943ea88e1f6dda4b : SUCCESS in 13m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/437ec3975b754ef6bb9903653e4cff4d : SUCCESS in 10m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/29cf26fbdb324698916357d1c536b19d : SUCCESS in 16m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/309ed034245e42c0b6b4b774f60f94eb : SUCCESS in 11m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f5c5b4bcf483497b8747b7d87031190a : SUCCESS in 9m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/67a6836307ec49a8b69d937415674065 : SUCCESS in 25m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dedfaa05fa4540c38faa19c329fde93d : SUCCESS in 15m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b065090a7c64e06b7a1454475282f6f : SUCCESS in 5m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7910b7570bef4e0f8f638fb134482750 : SUCCESS in 18m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e9920101e0d0414997ad488269dab8c2 : SUCCESS in 7m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad16e9ec568444f0a1fde8720c96ce7c : SUCCESS in 9m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ee150e5c92d94726828f0e7d3c44abc0 : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/995605fcf8484d2eb7a2f545cb1b54df : SUCCESS in 47m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac3f602d754b4a258ede94d63cd18b1f : SUCCESS in 1h 21m 38s","accounts_in_message":[],"_revision_number":1},{"id":"29135a31bb2f96222a01db7f889b0fa47c5547aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:39:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f095959d5dd474fc96c1227db415d449ff9e0f1a","revisions":{"f095959d5dd474fc96c1227db415d449ff9e0f1a":{"kind":"REWORK","_number":1,"created":"2021-08-19 12:49:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/05/805205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/805205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/805205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/805205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/805205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/805205/1"}}},"commit":{"parents":[{"commit":"ad5b9d5625ae3108d04f9cbeafc75363cbbd039c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad5b9d5625ae3108d04f9cbeafc75363cbbd039c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:49:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:49:05.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 2.6.1","message":"update constraint for oslo.privsep to new release 2.6.1\n\nmeta: version: 2.6.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Id7ebd54c7d585550fff0e3e78a4ba3b9449dcd9c\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I2b9b9edf5384ae66ed1e81c817d3e7b12841b00b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f095959d5dd474fc96c1227db415d449ff9e0f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f095959d5dd474fc96c1227db415d449ff9e0f1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
