)]}'
{"id":"openstack%2Frequirements~805211","triplet_id":"openstack%2Frequirements~master~If9877299bee2ec4a27049fd6f7bd2d6ba6720abd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If9877299bee2ec4a27049fd6f7bd2d6ba6720abd","subject":"update constraint for oslo.policy to new release 3.8.2","status":"MERGED","created":"2021-08-19 12:58:07.000000000","updated":"2021-08-19 19:31:11.000000000","submitted":"2021-08-19 19:28:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805211-new-release","meta_rev_id":"5787e38d8994f71df77af7476c36fabe9424d752","_number":805211,"virtual_id_number":805211,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 19:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-19 15:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-19 13:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-19 15:19:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 13:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 13:31:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-19 14:21:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-19 15:19:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e0a3990c9efe018402943d9aba378d44ba78c645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-19 12:58:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a569e089908b66cb92d4747c2636c138e5265c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:14:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/58f5c375f6bd40cd89408bfce80e3c6d : FAILURE in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"04c5c70000f15e909445beea53a7e1df1c6b1485","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 13:31:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"532b1381b9edfcc1c2ddd6ea5e52b28975e7dc87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 14:21:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8321d7e0a9854b2ab5b86bd35dafaa81 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6b9e53e27f441bf8499ee7c6d5193c9 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acf79aac0d774e2aa97d1be9d6e709b4 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c32b5b041214d7cabc43aa90a93f351 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c5eb9606f4744feaca266d67a309d01 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55bfd1e68024425bb3f7ca428e42e199 : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d14790a55aa4809ac82fa4c409a6d4b : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f748d8175214620be09fb9ac9370b69 : SUCCESS in 10m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3aaf7bbd556c477e9bfea59a62bd31d5 : SUCCESS in 9m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fc067229bbfd428bb69cd4bc95c402a6 : SUCCESS in 6m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7613e7305bce4360bc84e9055611a204 : SUCCESS in 9m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/125a07329d264e7089a477c92fff2d47 : SUCCESS in 6m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/93d4b5ce81244181b4c5f83e25c2f0b8 : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/427a220603634b51abfaae54191e08cf : SUCCESS in 9m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75ffe0633ba6431eb748f69576b122b6 : SUCCESS in 9m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/26186aab0a404dd8a8e82ceea19f8e64 : SUCCESS in 14m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f4734fba4afc4927b5f58418f5002194 : SUCCESS in 10m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/083ad2e5fbcb4826be5198f6171cbda5 : SUCCESS in 5m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4f4b330012fc464ea3635815f5f2fdb9 : SUCCESS in 23m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/00919d3fbba142368e55a2151f31494e : SUCCESS in 12m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a64e915727d842b68628ef20dec5f439 : SUCCESS in 6m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/206a4be088e64cad8395c5cff80c7027 : SUCCESS in 17m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4b61d3e55056479f8b4c5f6b13459435 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ef3f4be5520462ea2e2e4054db99cde : SUCCESS in 7m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/30576f477ea445cea0469148aaeaf906 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42c0ef3da7874184979f698728ae23b6 : SUCCESS in 49m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46b9cfbe390844f488e264a844630442 : SUCCESS in 1h 17m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6372d3adcbc5876202438e242e7089e2d300ca8f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-19 15:19:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d98e5a2aca91627cc3af77d974eae49905346c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:03:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5e2e3c75c763beb34ff69ab912c4b5f84f6b5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 19:28:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1f803908624f2b98effb35edaedc12 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f8881c499f34f9e84bedbc94710c1e0 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/334bea00cb6b48dba00d572e7b709ccf : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/410a7b1853174a11889e0ed6118543fc : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2576140d37e9400c869906b5be255ad3 : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/834b0b02744e4e2297d89828e70af38b : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c85044ee875e4cf78b042667c7ebee8f : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f39bd4a59304de0b6a029e562fdea83 : SUCCESS in 9m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/16617797217e4c729cb4afb2982793ed : SUCCESS in 10m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c6c727813f854c8686ba25de49dc3b09 : SUCCESS in 7m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/37575c7157074e0e8b34f2e963812e4c : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e56aac72acf411f9187147da4f0dbbf : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cdecd6b4547a4b258ede51cd745ed306 : SUCCESS in 9m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9d08c06ab46941be82181b413b7cc4cd : SUCCESS in 16m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5e9b62dd397140dea5d3914d61b26e08 : SUCCESS in 10m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b7959730c297400d9e42b92c202a4f29 : SUCCESS in 6m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9e18657a4f29414db8b0ff1a5b8768e0 : SUCCESS in 25m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/45a6f25c52ed499ea50bcd38491dc3c4 : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c73480cf0f66457b886eadbf94f1a100 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f66c97d467f54b26aaeb1b44e29bfc76 : SUCCESS in 19m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9e43c13e7481495d999e1f2ee200b357 : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49f288da8ea34ef0a6c9b75bde4f0cca : SUCCESS in 8m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/22559cb5ef364db79c06a542c4baaf42 : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60cade2ee0194113abd03caf3384a315 : SUCCESS in 48m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9591860a0d1b4cba8c957980910bc83b : SUCCESS in 1h 12m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0a03d972f8d5f733a2b09c3be4ac63f140702b1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 19:28:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5787e38d8994f71df77af7476c36fabe9424d752","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 19:31:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/185f0bb468704dfab0b11c24f28d0f59 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bc4479d84e0a383ea6b6ccf696e931df21e7b0b","revisions":{"8bc4479d84e0a383ea6b6ccf696e931df21e7b0b":{"kind":"REWORK","_number":1,"created":"2021-08-19 12:58:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/805211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/805211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/805211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/805211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/805211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/805211/1"}}},"commit":{"parents":[{"commit":"ad5b9d5625ae3108d04f9cbeafc75363cbbd039c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad5b9d5625ae3108d04f9cbeafc75363cbbd039c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:58:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 12:58:05.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.8.2","message":"update constraint for oslo.policy to new release 3.8.2\n\nmeta: version: 3.8.2\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I2d54646ddc5659b9a89c1adb6bf635f50871d9e4\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If9877299bee2ec4a27049fd6f7bd2d6ba6720abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8bc4479d84e0a383ea6b6ccf696e931df21e7b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8bc4479d84e0a383ea6b6ccf696e931df21e7b0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
