)]}'
{"id":"openstack%2Frequirements~805220","triplet_id":"openstack%2Frequirements~master~I78f9c066d853def41d35a6db536922f9bd49f392","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I78f9c066d853def41d35a6db536922f9bd49f392","subject":"update constraint for glance_store to new release 2.7.0","status":"MERGED","created":"2021-08-19 13:27:39.000000000","updated":"2021-08-20 17:22:44.000000000","submitted":"2021-08-20 17:18:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805220-new-release","meta_rev_id":"f6ee6ef143f914a47b27cbc4f92b22df64ac7c5e","_number":805220,"virtual_id_number":805220,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 17:18:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-19 15:19:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-19 13:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-19 15:19:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 13:30:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-19 13:51:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-19 15:19:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-08-19 15:20:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d43af82c00b79b9a0cda45af92dff52605b9e11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-19 13:27:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"116e4158dd791b692eaa638ce3f8cb91765c9342","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-19 13:30:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4544ddbe11669a41782c36d1376ad871887d919","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 13:51:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ffaf5a236afd4755940543e0678509f8 : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a4b5e8dbc03ef403f958d31ce4a46dc013cc5176","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-19 15:19:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e8202f6c8510fc6d9d74a54059d0bac3927c93e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:20:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b085929d8a4966ab4434207c86d801 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb3323b74f74070bae8b415eb71b352 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6f08fde99424a5088637c61b9670099 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc9ee2f393e54a05be5a846c2f2cc349 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74f2e7836ef141e39f338ac4f7152f3f : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba96df9c953944e79306f424a4e39dc2 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fbe2a28c8ca4fdfa764d3246212e927 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc3bc07e3e1343a49fbb51a45ee0150c : SUCCESS in 11m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2d2ae1937ddb47f2bb8d4fb89463b73a : SUCCESS in 9m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/34f22f52b38c427a83d6921dd0a6259f : SUCCESS in 7m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/28c336d7bcd341469bb2883ae5a9f1f1 : SUCCESS in 10m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d1ead43c92f146b991a4d67a9562205d : SUCCESS in 7m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e027d8653c4148d88562b29c19bd3942 : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78d29e52cbdf4596abe280023ef669d1 : SUCCESS in 10m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f6c0829310b345d2ae9c7eac2b858143 : SUCCESS in 10m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8518b7aad4a048d7b8c35e731c91d0a8 : SUCCESS in 16m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/abe50e08cf214320851c53157e327e16 : SUCCESS in 11m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b7503b5fe0ad4f7ba8d43609db6f4c21 : SUCCESS in 6m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d9a091a6914c409ca54e8b8fd963d4d0 : SUCCESS in 25m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e1eddebe007244d1b77c4272cd6890d8 : SUCCESS in 12m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e73a363a80440dd80a66a15680cb6ba : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f48948f78467416596cbf764dd442cfb : SUCCESS in 18m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e94ce0adaf8041e1a568daa2258a5cc9 : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de0cce0ee1574d04b8b6d1d7e820bbcc : SUCCESS in 8m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5fef737b8a244dd6a7288cdecd77c676 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ed7f5731c394474b6920e76c2c5a86e : SUCCESS in 48m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4f3723280264857ad56f3e3653d7c8c : SUCCESS in 1h 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"af7800dbff76e5335156f9f05eb3d611a78296d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 18:14:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edb23b3815899be8fd93beaa3bff7d8192bba162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 19:39:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72bcb2505be04850be8d93f22cebcaa9 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f0162c5fca4f6783964f3564bd29fe : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09dbd4fecc9a4c0e941f25a80cc1ad38 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee14bc6c308045269c49e04c496e8a5f : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c0d989e721e4ae49f8cfe8d632b852e : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4028ec774bc04074a88b9e7fcaf418d8 : SUCCESS in 5m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bbd60acf1a814dfd9814cb51b5f399eb : SUCCESS in 10m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a9435df61014b2e9af9177afc69ecc7 : SUCCESS in 10m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/11f846dfa42d429d9bc4b8f9bcd13909 : SUCCESS in 10m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3e3da3a8c9b8450aaa2feca137ef70a0 : SUCCESS in 7m 30s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a96f4df3793740c0890e58a46d2b00d5 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a15fa2ca881245fba59ef4ecabbaa3ea : FAILURE in 12m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8d987ebd932b40a790af378381a0cb6a : SUCCESS in 9m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/99497bf8b39e4395874be45d1d102d41 : SUCCESS in 17m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b7de09eb3cf24269985cd84f075e8f1e : SUCCESS in 13m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e76abeaeae4e43f09335f11e7b429dd7 : SUCCESS in 7m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4f9dfd39695644779287bf1fa88e10a3 : SUCCESS in 22m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/57d57521dd534a52adb99e91cbf67b61 : SUCCESS in 13m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d4aefbe430648398f5e0a827727fa7c : SUCCESS in 5m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/870ae94037e241a191146ba7e4af39a7 : SUCCESS in 17m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4d3917980f8f40f0b16762ac0584e9cb : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/347870329a0c44cb9720290a7e33a819 : SUCCESS in 9m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/867bdaec1b544c1b87a183196927b13a : SUCCESS in 16m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7f3b030324d4972bbd87263b851724c : SUCCESS in 47m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7c1b5769131479d9a7f6a3124770c81 : SUCCESS in 1h 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fd76224c0023fdbca1e78b3b20ad9842a8584a53","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-19 19:40:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f006cb82d855d27f41496f11709a8edeeddf4daf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 19:58:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e5d32dfa15c74fc4b7415bbf6497b62d : FAILURE in 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"627cf69ee921e774d1a6a71d57bf5088fa9c7d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 20:45:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8a07b32a68435d929f99918c15b483 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451deb0770c84dd3973dc4d2377c14f2 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1633654b81c459cbf84a73ba1e1a18a : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb76645ffbef4cdaa561d22491463577 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/640fd302338d41a99ec576f8d175a0ad : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fc5876d45fc4c92a31031f21be8fd46 : SUCCESS in 6m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7447b1aabee64612a63cbf01f9902f7a : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/20c732b337e14d879bd32f2731c8b3bf : FAILURE in 10m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dc23232f172d4624800250d2771a3ecd : SUCCESS in 8m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/81b3696cdb4843819608adc93beb83d6 : SUCCESS in 7m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8b9cc1c517af429d9d1c21b3ea76e112 : SUCCESS in 10m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6f1d34f891ba4b659fe9e84c56c8fa48 : FAILURE in 5m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/869f075569f742f3b712b3cb88fae1fb : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03f43d30685d40c8ba9e90c29526e112 : SUCCESS in 12m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/802e79d3f7bd4e48a5768a55af1cf088 : SUCCESS in 7m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b02905f827544e6daa6bad9b6b9a3236 : SUCCESS in 16m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eb0e3b7582cc420794e23442578c3fad : SUCCESS in 9m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a4336d5676af4da89e318e52c4530f9d : SUCCESS in 8m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c2acbc1d79e64fb495b661d70b33a0ef : SUCCESS in 22m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a2cb09e6df444a5d80109be880a1d42e : SUCCESS in 15m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e857a998118d472581e98ef1c683967d : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/925f83842acf4c939aaac1febefa32e9 : SUCCESS in 20m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f09503e9934d42588ebc302a453ff7d7 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/294f2217c7704f9c84ea7ea759158ff6 : SUCCESS in 9m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f4381ef02e4c4c819457eb435851f92b : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3782abb278d441099854a3ddffea575 : POST_FAILURE in 19m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5600de32700f4be09c9e75b461ed1a49 : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":1},{"id":"8e51725fd573f04c91d6aa85e1194fb4120efce7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-19 23:47:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"309f99b4552cf92f21b0b6e24742f24b8b55d8da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 00:03:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bd8d1c7a1eab41adb9536f71ff50824f : FAILURE in 15m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8f41015294b92ade991312d6819ad9136d0fb86d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 01:09:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a802040e18b45928e611f39c780084d : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f70ce488beb443199a3131ad88b93268 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/248307618a254641bcf66726a416ea7c : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33c361aa3a04c3eae1a32f8da2b48bc : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5189fb008214cccb928e23c731ba4d3 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8722faa6c5f480687813a8c07fe9f07 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce63f5c9047b45baa0d2db9d0d4c277d : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/84ef911cff8b45eb973ebf001213bea5 : SUCCESS in 9m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ce57abeebf3459ea78dcaf0af9ef062 : SUCCESS in 9m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5d6da267cf854135a80df037f672ee36 : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/64aa38dc739c4b02b91833e63c74d7a4 : SUCCESS in 8m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dc290bbb3f494886b43f6746f8a45555 : SUCCESS in 5m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/02f55e639cc04a37b9e790ba4a1a1231 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f94e87a8e3f4d18a56c8b6b9a9021a4 : SUCCESS in 8m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d33f38c2aa764bdc878fbc0033db5a97 : SUCCESS in 8m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/19514fe7f35c45f48b90d266811100a0 : SUCCESS in 15m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/afc99bf3adae4786abaf380d13bd5b5f : SUCCESS in 8m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cd4841e96c7049e9b3f27c5713f13565 : SUCCESS in 5m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/48444e44b62c40af8d7d4ecb0cd50df7 : SUCCESS in 23m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/10cbab4c250c448a8e8b0a6d846c9cef : SUCCESS in 12m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/06e7342ac58e4791bdf9eb19fd17c63a : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e93120420e7484fbb41f08a3d64b3ba : SUCCESS in 17m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6315b6fe1d64486a907c63e97f3c4208 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c5b919c724b44edc872ce0ab5df93c74 : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/32432a06235345d890b178b96128070b : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbea356dd2d64558a540dfd4e5cd7034 : SUCCESS in 43m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e2042c7cf124128a60c66dbbbf4c7a8 : SUCCESS in 1h 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"630dd822a8259f610d37d924106742c947c1ebb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 01:09:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1035a47777bb08c9a3341d8d8d192ca613bf39e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 02:24:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a85e1c3b724ea5b02286f79066437e : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0bfdbcf3d949bcaf1726ab2b2bac6f : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/456d26defb974e65aec813b9dbf8390c : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4415ca2508b4bd8a25e02aff7586cc8 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fee10ef73763485496bd8e6f05ea42ff : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2604828c199e4d3f914a9e68877345d9 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9325d1f9fe004264be07329bd141a0ec : SUCCESS in 10m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3c21199b404a46ae82e7990e6d2d6927 : FAILURE in 5m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fef78c3378b645c29d59d06f946b67a1 : FAILURE in 5m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7a7c91a938b243fbbb7ca4e776aa659f : SUCCESS in 6m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/18484022e32f48b6b69dd199ee0a745b : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18fef98b6f5e4c70a804e656de7c910a : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/731fafb38e154abd86e8d0e92bdf1fed : SUCCESS in 7m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5e66ffba6a02433a8bf89d9ef2ab1b08 : SUCCESS in 14m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7455006eb047418f90873470f2defda5 : SUCCESS in 8m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7df6e864a0f4473281e75a9dae9e2a2a : SUCCESS in 5m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3e9ce987ae804fd2904c8d52f2a9da04 : SUCCESS in 20m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a2781130ef0a45b480082dbdc4e28fea : SUCCESS in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/240401051c974edbb97c05d62c838a36 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31bd2c374850471ab24165b6d9a62b77 : SUCCESS in 20m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c8699d0b57c14e04890d405debbfe1eb : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c157ba3155a64d6496b5bd5ce314f865 : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3066b1cd64d54d7390bc48efdb2d6721 : FAILURE in 4m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c75395282e344fe83ac5aa5d13c50df : SUCCESS in 44m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0efe83ae24d45ccaa673c3a2fd85a25 : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ad049822599b8c97e32cdb4a2b7dc4c6763725c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-20 03:01:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d04d23143613eaf21eeeb6d0eca34731e1c7b3c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 03:16:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1e795707cb734550b396ad086cee7bcd : FAILURE in 11m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6087ea4761a868afb1594518dcfc595e22115e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 04:31:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22520840eea24905af1898bed168a0e7 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df310e10ecb4babae7cc712dd2ba5c1 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/549889074fa14ed2913c6a356eb3c61c : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b1876e075fb4e1d9be6266974afce0e : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35218d39781a4f699e7ba5e94420ffc7 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de545d5c572f482ea68e49dcb141d54d : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2bcb5bc31b14e5ba8d2122e777efef6 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9a86a2b2458141bcb6fcd64504fbf28a : SUCCESS in 10m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b999670762b64b8380c16a9ec149fb64 : SUCCESS in 9m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/197ccd1d9fba4098a89e50d4251a5d4f : SUCCESS in 6m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b665b294033348f0bb6410cc9fd2a102 : SUCCESS in 9m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0189702e4e8c4866a33329c2ce77e3a8 : SUCCESS in 5m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3d76e1eabc0a4cd7b9ddb3c94931634e : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0be929684aef4d99b9c04f88ff18efb2 : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e88ebeaf59f8409589664f44f33a847d : SUCCESS in 8m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/065e94d956c942d1b5a692ecfba20d3e : SUCCESS in 13m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7748990ffa854097a2e3dbb412886dc7 : SUCCESS in 9m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/72b593e6538649abbddfbf698857c1ae : SUCCESS in 5m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6251ca28b37c420690e7dbbc6244cb19 : SUCCESS in 24m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/672da844a8fc4369a6ca0addbedd23d5 : SUCCESS in 12m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/33ba07e2249443df804464cea5d3a569 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ae15a40f1a9043a4ae15288dd2e23bee : SUCCESS in 16m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cd9ec0d20d6d480e8dfdaad198f81b34 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e19d377dfed42fab9e0e13ba8b7e43d : SUCCESS in 7m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bc44faf6bfe648c58eb8df150747cf68 : FAILURE in 4m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9ec6225561a4f83aeb7214fd079c3e6 : POST_FAILURE in 14m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55efa2bb28494087ac307c6a8f1c8093 : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c0c98d6010ae3305d274dc61639acd1e51589e19","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-20 04:47:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b744793620c57cdef1266a74dfcf2e554b25838","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 05:04:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ddf755ff22ae4ff487e56e02273f9ad5 : FAILURE in 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0646aa8be7b6a39c0650f79060fc3c9c690bc017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 05:35:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eebf680428774d9581fb098e01d35225 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78c02971f0c4d6597af4ebadd14ad21 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a00b309f6b7f42c083b09f8b7f73eb70 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24e85c5bdda747d0a1665d4800200c3b : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06486757ca3146fa9b28983f8fcb82e6 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2aaa3aef94064484b6727c3a176d5b5c : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a028ea9a6854088a63711b568ae218f : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e92d277e70424306b13a3c851f877ba8 : SUCCESS in 9m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/991d1e1bd08a46acac6c1a7864c9273d : FAILURE in 5m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ed8390625cc74908942a6c84e82d026d : SUCCESS in 6m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/30ab3fb531a940cb86ce60645fd15d66 : SUCCESS in 11m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6ddde6179b6e43a49bddec7bcf144b75 : FAILURE in 4m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ff3f72fae79747338e2ec2651488abf1 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01ac827e75ee4238b2250d31688c47fa : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8e237aeb7f7449b5870644ad38b130c2 : SUCCESS in 8m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ab1db7baa63c4c7a849eb046c933aefb : SUCCESS in 13m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2fb4e76ab532408483d44e91dbf473fd : SUCCESS in 9m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/160eae50644b4183bb8088f9aa48fce3 : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/994d21f0148e4879a01485c70a84455f : SUCCESS in 23m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5d3a815a5cc54684a0f65b695ad2b58e : SUCCESS in 10m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2431cf40e0c24059850c44e1c5e049ae : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c57e978a668b4ce3bb5cd9f0ca7d20f2 : SUCCESS in 18m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4f85b4d306f64c6a8577232426638d14 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/afae017cd9b6444cb29d5fd83de3f4c7 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f1bfb44889cd49628d0fa8425d4f86ea : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/765a1fd59bae47df84a13d1b3a9929cd : SUCCESS in 44m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a382d3a3775f4259a5aa871129b0ba05 : FAILURE in 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bf127828b01d592e61757310077e75473f122caa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-20 13:36:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"13fd520d56adc4e32b8f30a72ba7b62b2e33974c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 13:54:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fca5fb23e8324879a6e9cfab27123742 : FAILURE in 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4916403625f63210951df1ac72304c22803d9575","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-20 13:54:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a741004d9279074aa96e8813f7f66d3422221ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:17:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1c9dd3ff70f54ef9aa4a70ea6d20a2ba : FAILURE in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"bf566c113d89fd1eb608ed40dfc61bf8dbdcf863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 15:02:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28847adbfd8540ef93b90c8c1b2ce8d7 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4bd116dae042249b5a2f7cbe6aa0c6 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/417dfd5a33914163b7532b0d5807bacc : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7151252bf64e4e8ebb514470bc0db9b6 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/166ea1611fdc4736b978390672ba4fa7 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b6fe4c47a28494699155fc675ac08a8 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7c30c081d924dd0943594dbb169e9bb : SUCCESS in 6m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3c33b1e102bd4c878fa9b9c68b1e928e : SUCCESS in 12m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4ff325921edf4ae688c78403bec88efb : SUCCESS in 12m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dc461005c11946fa84a4003a09e4dd97 : SUCCESS in 7m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a8e3aab20c6a4f4a9e649d0c0fbf7723 : SUCCESS in 10m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6564e8e26815441789766d3f03549a9b : SUCCESS in 6m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/582bc2e782464788807f9da8c77ed692 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ae9956a6a354664bc12eb6551095573 : SUCCESS in 10m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bab144ec965b4a478afa4ebe638066c2 : SUCCESS in 9m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/53569865c3f74b53bd9bc6ed9f06b750 : SUCCESS in 16m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a12326895faf4819be312111af826381 : SUCCESS in 11m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ec7d98a75b254fcea1a9f763555ce287 : SUCCESS in 6m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c6fc130422b34209afd6ad054ffdc202 : SUCCESS in 26m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c325d8406e7541a99b671aaa3d6eed5f : SUCCESS in 14m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea431977a0304b68927786d738b194bf : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5006d2d15f4d425899ac044fdc410ee3 : SUCCESS in 19m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9f47e2e8325f4b7db58c2327799a9c3c : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b30da9635f0b47d2aa696ac343402dd1 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f27d5c25df73405cb208a0a9b702bab8 : SUCCESS in 14m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36c76727acd4480c9e5664120a0f528e : SUCCESS in 48m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2e9e2f74b6f47cf8a2de00be33a9dce : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c0fe3c267b810df6e36c8fe8432933dc64c89da3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 15:03:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57fd0f38f4e0bc67a3406d8d1fd653c6d0eb1a27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:18:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d729983fe0c4f74a15a0aafd78769e5 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058d3e10552c40fea7fa2c21f7b18067 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71ad6a438806466684ba27d2597dbef3 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afe583e812024a5db1b7c8c73841dcb6 : SUCCESS in 8m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/889505cd22cb4bf4a0bbb01da2459bb9 : SUCCESS in 6m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44945cfaa3624768a0f111bde18d0ce2 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e3f751c47004667aaf29a7c4636df76 : SUCCESS in 12m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b45e1869da274215881cb74fe1d749a2 : SUCCESS in 12m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8b3f30fa304548ed9eeb18aafa282fef : SUCCESS in 12m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f3b835d896a54062a6fe3137ce63cca8 : SUCCESS in 7m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/aa35883d05cd46dab8841a5139f8508c : SUCCESS in 16m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea04fad75e3c4ca6b7daab4b841267da : SUCCESS in 11m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69f4bc70b6764603b3c940dce3fa60ee : SUCCESS in 11m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/37481a9315a548d182f660add639bc17 : SUCCESS in 15m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7de537942ffc430ba9258bb6dbb51f3b : SUCCESS in 11m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c37057ef6ae04f548f07034a1a220f23 : SUCCESS in 8m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/625aad449fa04d2e87c8c08711a270fd : SUCCESS in 29m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ae69424b1282431894de11283e3395bb : SUCCESS in 14m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/413dd1557ffe4cb08a1908f61327a990 : SUCCESS in 7m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef16a0564c1e42adaf2ae423570e92fe : SUCCESS in 19m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b0a6162e50114294b3ef10473b0edd72 : SUCCESS in 8m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/396491699c794cfe8a034118167a2de5 : SUCCESS in 9m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4db2ee3fc02640a6ae8dd58204cc98b5 : SUCCESS in 14m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcf6b6bd4f7e480faf01688a3bf05e48 : SUCCESS in 43m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eeda684278684de5b0c8bc104a404e1b : SUCCESS in 1h 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1b078a0acef98376ce90b7c31089289b848e359b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:18:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f6ee6ef143f914a47b27cbc4f92b22df64ac7c5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:22:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc02704c5f449b7ba9bcc9c1b00d875 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"351976550c476c1fb24ae85f2947438dee918ce1","revisions":{"351976550c476c1fb24ae85f2947438dee918ce1":{"kind":"REWORK","_number":1,"created":"2021-08-19 13:27:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/805220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/805220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/805220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/805220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/805220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/805220/1"}}},"commit":{"parents":[{"commit":"ad5b9d5625ae3108d04f9cbeafc75363cbbd039c","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ad5b9d5625ae3108d04f9cbeafc75363cbbd039c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 13:27:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-19 13:27:39.000000000","tz":0},"subject":"update constraint for glance_store to new release 2.7.0","message":"update constraint for glance_store to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ib13cbe6e35c83d89216ad24958c4004affba13cb\nmeta: release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I78f9c066d853def41d35a6db536922f9bd49f392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/351976550c476c1fb24ae85f2947438dee918ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/351976550c476c1fb24ae85f2947438dee918ce1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
