)]}'
{"id":"openstack%2Frequirements~805339","triplet_id":"openstack%2Frequirements~master~I0c2e2bc0bc67a7f1f6d12fdc60c96b46d12e0fb3","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I0c2e2bc0bc67a7f1f6d12fdc60c96b46d12e0fb3","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-08-20 06:34:50.000000000","updated":"2021-08-20 23:13:32.000000000","submitted":"2021-08-20 23:12:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805339-openstack/requirements/constraints/noclob","meta_rev_id":"c471d174cfa2010d8f0cb89fa049319bc9bf1e20","_number":805339,"virtual_id_number":805339,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-20 23:12:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-20 15:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-20 15:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 06:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-20 08:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 15:29:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b3518dc11873020832f0de63aa86d784e1859f3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-20 06:34:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02239b1271fcedf548e6d7d8b31cdb0fe5e41b68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 06:47:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a70ead3c03ef481ca6e9582538c988c2 : FAILURE in 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"aa3072cc2a5666a67288189ec69784bc34848b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 08:29:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a79b05aacd4bc69e60e98a7753905e : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6cb905c0644290b6fd3f6a442e6a79 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1d73515f7c94696ae82302671bedefc : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04e1505caa7942e4b28e20e333601348 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2227dc691049499a8de87b43966beeec : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b3e024d29364ab0a2aca8299f00d874 : FAILURE in 8m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/09c024597eae468c903661df6a25a5ba : SUCCESS in 6m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/52fea79dc8244065881f31b7750838c9 : FAILURE in 4m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56a846b085ec470ca1457e7f2076bf6f : FAILURE in 5m 58s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/74300b76452641b1b8235a0739cda3b4 : SUCCESS in 7m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fe67c3f20ae44963a11b594a809b2ee5 : SUCCESS in 9m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/09e8f10157024c979bbc081228607b42 : SUCCESS in 7m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ed2dbd613c8e4d8dae7ffae495435480 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c95b56ba839f44948f3f2a1991521243 : FAILURE in 11m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fb8d68569ac94f35a5fa998c184fbb9e : SUCCESS in 10m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/242d168906024735a9c9dd17934f7b8d : SUCCESS in 26m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5c45cf6037694ef2b18ad6b44195ae8a : SUCCESS in 12m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/faa9695fc6c443ce93d1b335ae991606 : SUCCESS in 9m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c1e9ed53317043c9b1253fd8935473a0 : SUCCESS in 32m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/073d13098a1f4b6cabe18448d56a8c5e : FAILURE in 13m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9744005005914a999c85a6ca2cdd96aa : SUCCESS in 5m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f39ef88387bf4e36a5a04183a11d0066 : SUCCESS in 18m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f9e4551d2fb94cf2a2af922072c7c15b : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/688ca12ead734cd8a619b1d1595a3fb8 : SUCCESS in 8m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c1e9e65a2ffa4fb4bcc50a520be054f1 : SUCCESS in 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7e7d5dc37b245dba5cc0c2348a86228 : SUCCESS in 43m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ccac8c6a9f994f05b66e207c9d5b49f2 : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9bc82e6fd6c6a610b729b1061a09b688e7acadba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-20 15:29:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94870f346bf6963ceb2b9194314cca84b1ca54e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-20 15:29:07.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"d0bd07181161237d32039310f1457a4d91d0949a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-20 15:29:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fc16f9adc79a10bf0ad78fd1915b5f3aaefb62a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 15:46:42.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/48a727ce02ca470494aca09aa97288c5 : FAILURE in 15m 44s","accounts_in_message":[],"_revision_number":2},{"id":"60881deac944400c2d8475637209376007e53a49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:15:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa86eb85e74241dabde532e68ae9f42e : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528244c8575a430791d95908e546124f : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bda061cf7bf43f4b1c058fa407d280a : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85f1c2fb1d874cd78b9dc5d8369b580f : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/362f1e4b225445d6aab7540218e85dda : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfbaa3885518449d9cea4840fc426aac : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d64d4eaf44cd4c6e9593c0b5429f191e : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b6fa2c0463be476581891014603dd4af : SUCCESS in 10m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8022356e7f554a30a154695ff08b1f3b : SUCCESS in 9m 32s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b3f3d6f75f654ed7981829aba5892cc6 : SUCCESS in 7m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e849e03e951048cd9a26ebe212c76fc3 : SUCCESS in 10m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8ad5d9bf89ae4e59981e1b8079a6d25a : SUCCESS in 7m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e6933e247e694ff8af29d7a512402859 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ccb03baa6952420593dd6de9b8226b00 : SUCCESS in 8m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/33fc41a9398046db98c90967caf6df43 : SUCCESS in 9m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/30212227663042afa40b895dec9dc3e1 : SUCCESS in 20m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1e93ec1f15204f029665a87539949457 : SUCCESS in 10m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/414d2afd7e66478cb32f526252c463f5 : SUCCESS in 7m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fcd1256af98544608526b0aed8ecb539 : SUCCESS in 27m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6d69e4486b994f7091d0a5f0b8456adf : SUCCESS in 13m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7282d4f86e6940a8a52a4c62c0fb1a0f : SUCCESS in 5m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0152da0c5e0e42f88804a81ba0524ab7 : SUCCESS in 22m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d135094c2a1140deb0ef8a0e0b6606f5 : SUCCESS in 6m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3065bdccb6aa4cf0b5b653c20425c96a : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a835338853d64dedb89ee5f9066b8347 : SUCCESS in 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/649aabfb7ff844f184611c39f09e71c1 : SUCCESS in 43m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/403b016df40d48288e68e86677b43c2a : SUCCESS in 1h 37m 08s","accounts_in_message":[],"_revision_number":2},{"id":"df641d28fcf332c02488becffd7f2ba4fdc546c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 17:21:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8af78dd625e03e2eb27c775b214d28360f9c5f52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 20:18:23.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac627edb6716fe64a99607764645e2644e1cb123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 23:12:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad52a3716915453abfd70d4e99806b7a : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157ec60ac5e64bc197159620fa4a01c4 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d411701f04143dc83e21f66233aab8d : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adc8a55eb7c94f2db239fe9eee228213 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ecc7eec9deaf418db288a6b6f823247c : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ea6350182260412d977ada41a8e0d7fe : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c74d41bd7e674e45a2b6cbf648bdddf4 : SUCCESS in 11m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f09009af533341b9a6a25cb2dfccc647 : SUCCESS in 11m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cf7aaddbf010460aa5e4d31d773a9fe8 : SUCCESS in 10m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/27285d5f7e2a4d76bcaa50bbe6d70d3d : SUCCESS in 7m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/361609f6ac1243fb973855fb10f0d835 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f354922ca2ac4b44851ed9d2f1afbb6c : SUCCESS in 9m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2869a558085349a7a970aafb2d71ffa1 : SUCCESS in 7m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8229c2161f064055b0375efe35280b22 : SUCCESS in 14m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7bbac4487e3b41bb9e885b443fcbd637 : SUCCESS in 12m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/863e559242a94be898c6f66ec815843e : SUCCESS in 6m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8ea91028663b48bd9d3e564b3f7ddf16 : SUCCESS in 21m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/026fb2e8e473460ea976955c87854697 : SUCCESS in 12m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f9481abc97849a0a76be3d719a4d5dc : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e11dd6b37814b6d8711396e2f06972e : SUCCESS in 27m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0839bd6086534841855c00a66afd92ba : SUCCESS in 7m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0f32b8250b36475fa72b4d9c69385b9b : SUCCESS in 9m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/612f31f9970b4bb9b563cbec52be5797 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a3b72fe69fd44d3803be5e5cdec375c : SUCCESS in 44m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2070db0e364438191e7b81272c05df4 : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e39c0a11b8911073c38e2480b2958cf05daf3232","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 23:12:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c471d174cfa2010d8f0cb89fa049319bc9bf1e20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 23:13:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c3b27732fe54aae92cbe5d407df754c : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72a895f4c020dd97c74d97b8dcbe2247dfd7bc3e","revisions":{"c1b0321b9064527f15250b1b53367eda5850d0e6":{"kind":"REWORK","_number":1,"created":"2021-08-20 06:34:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/805339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/805339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/805339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/805339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/805339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/805339/1"}}},"commit":{"parents":[{"commit":"95e3c140452e42eb55b65f5dc3db6a524ede318e","subject":"Merge \"update constraint for os_vif to new release 2.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95e3c140452e42eb55b65f5dc3db6a524ede318e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-20 06:34:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-20 06:34:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0c2e2bc0bc67a7f1f6d12fdc60c96b46d12e0fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1b0321b9064527f15250b1b53367eda5850d0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c1b0321b9064527f15250b1b53367eda5850d0e6"}]},"branch":"refs/heads/master"},"72a895f4c020dd97c74d97b8dcbe2247dfd7bc3e":{"kind":"REWORK","_number":2,"created":"2021-08-20 15:29:04.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/39/805339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/805339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/805339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/805339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/805339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/805339/2"}}},"commit":{"parents":[{"commit":"95e3c140452e42eb55b65f5dc3db6a524ede318e","subject":"Merge \"update constraint for os_vif to new release 2.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95e3c140452e42eb55b65f5dc3db6a524ede318e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-20 06:34:48.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-08-20 15:28:50.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0c2e2bc0bc67a7f1f6d12fdc60c96b46d12e0fb3\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72a895f4c020dd97c74d97b8dcbe2247dfd7bc3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72a895f4c020dd97c74d97b8dcbe2247dfd7bc3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
