)]}'
{"id":"openstack%2Frequirements~805352","triplet_id":"openstack%2Frequirements~master~Ic8bf060742b360d5905ec3d93a386b9f6bfc2600","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic8bf060742b360d5905ec3d93a386b9f6bfc2600","subject":"update constraint for neutron-lib to new release 2.14.0","status":"MERGED","created":"2021-08-20 08:18:32.000000000","updated":"2021-08-25 10:17:42.000000000","submitted":"2021-08-25 10:16:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"805352-new-release","meta_rev_id":"fff7f786dcbe33ebe9708ca3c6836c4935cb3ab0","_number":805352,"virtual_id_number":805352,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 10:16:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-24 14:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-23 13:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-24 14:25:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 08:33:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-20 09:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 10:57:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-08-20 23:50:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-08-23 13:02:14.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2021-08-23 13:20:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-24 14:25:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e627c93b1d91ee15108202231208c836eae5e561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-20 08:18:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72a85496becba472da945a8f51c353f2e1bf13b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 08:33:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5a50b2068a154681bb80a5fee5124a9e : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"81e646ff609d42f9fe1bc10ee3dca8811dd2bb88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 09:40:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e9bdeeb729482c8766032429465a18 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf6e5beaace43bca4deabe85a332e44 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e0478c5269a4a19a90ae2a969e5a9cb : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/135d9fba22c34a6192bc7023a40f8c59 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a976f0b81f1e444f8d7d46e48c5c7794 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90f0c02ba979491da723e21ec5eb3bd6 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b6c42d431df440a850183035994ce4f : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4ee76c42b5d846baa42ae3995136bf0b : SUCCESS in 12m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/937087460a8d44d29fa018515f57c50d : SUCCESS in 11m 10s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/239f3a607ebf455c8f2b3e83d4794123 : SUCCESS in 6m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a95d5f92518246c4a5e631975ee15031 : SUCCESS in 9m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9ee977efec034cc7aca56885e2f36f4e : SUCCESS in 6m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ee7f53ac2d6c4aaa99bf0d988714b216 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e3244a7fdb044840a1f59b35c9e19c90 : SUCCESS in 10m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6b478129c003496b8613a38c2aab7f86 : SUCCESS in 8m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c8896106aeb645aba197c1199f3a7eb4 : SUCCESS in 16m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/789a0e6798504e298287483385bd37f4 : SUCCESS in 9m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fb1c0d16eb9d42a2819eeec0603d95b1 : SUCCESS in 5m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bacb8d9a45fd4ce592e11d2768eef7f1 : FAILURE in 25m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b3f5ecd5328246b7b17d527cd1482cb3 : SUCCESS in 12m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0058242735454e2581255bb87fbb9287 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9f7b0f19050490599f8cdc32464de11 : SUCCESS in 19m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b2e9eef884a64a9386fc337c33328289 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6de403ba96314bb4abe6bab69aa28cb0 : SUCCESS in 7m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/db53dc65ad5644c99d605847964cafe7 : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a69c56b125a645b2a6a51927d9d12f4f : POST_FAILURE in 15m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fa0020f97d84b238702c15a2bb14e49 : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0a59a8fae535df0fff4525a9ee7db73a02d3bca1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-20 10:57:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"236b99d72d8e75813d0c1d622c9eeaeddd8886e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 11:14:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c3944d65af5744ec9633b0820b5f58aa : FAILURE in 15m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7fe9b54f4ed620ee10bcb7afbf2d9f774d2ad4f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 12:22:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b8fae7b45c4f698e3fb90f08b88a58 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6cda4e69ce415899d4c2ab16ce6aba : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc50f130ddc4db3b944fcba5449fc4c : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c048690edac84bdb9ba7a2b2050b4e1a : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1024692ae2d3412cb4b78c72d87b8308 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5a952fba1b0442190757e4056fe072c : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af7b82777b2647a7b6159ed65881231a : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa892076d9334c83989fec42e0c9c66f : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2002f08dc21c406eac4988bdb1d7f162 : SUCCESS in 9m 14s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dc747c80f84644c390db3da0763525ba : SUCCESS in 5m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3263935fda6a418e9983c7d4f20a9f9b : SUCCESS in 10m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ae6b27d076d4460296554a439af63b60 : SUCCESS in 5m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b48389f2b3ca46b08f1619a534f23ab2 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63d02a38a73c4be9933712ee11492d26 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f8c3be4f9cae4d06a60a4bd206b26eea : SUCCESS in 8m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1027d712c85c4e05802dac1af9e2fd96 : SUCCESS in 13m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/52ab70a9d69946138e30a03f8a22c638 : SUCCESS in 9m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ce5786ac705f4c8ba2e66f3a647184ef : SUCCESS in 5m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4005e5f665194973b7c303360140bd00 : FAILURE in 23m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9e4921f71a004778a153f6705512aca2 : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eab03a1cd9db46c2915c20b8d5d72860 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec8b0214531c42258fb325c83119e42f : SUCCESS in 16m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1d127f2b99934b5b963856957fba2d23 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91bbd5da5ef04b6c8b37653ba5aa1d38 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/60b18bd22a5f4cadbbf87854381e316e : SUCCESS in 12m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7448756b3abd43a1a3d80f080aa2e3be : SUCCESS in 44m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ecfc04861ae4f85b109bb884e161119 : SUCCESS in 1h 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c47c741d7d6eddb0c351bd0714039be184d1608c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-20 23:50:21.000000000","message":"Patch Set 1:\n\nrecheck\n\nprobably legit","accounts_in_message":[],"_revision_number":1},{"id":"e4f2d50709a2c1b59f217002a55bfd7113685127","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 00:05:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/56b995479a5b4eb7921bb4d7d804389d : FAILURE in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"75e72d58dc6cd6b154832a396d2a66582064ec01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 01:08:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e245fb7fc2034e6fa5f7accf43ec743e : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd54ad36cd6f456dbf191e5bbf00dbae : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1ed6de4efc94bea8d066e76b00be422 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/124bb1b77c3b4f0c953a41f578050f40 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35a6b5280a1247e0963aad032b94b401 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64104f7b53264cc58e60e0ac716ee6ff : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7df19aba49994b0bbfc67290242eb13e : SUCCESS in 7m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/22b2767fca9e4a9181aeb66196b547fa : SUCCESS in 10m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cef8355c8aac459f9a084f8542961bba : SUCCESS in 14m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ed1678f5dd254e1dae396b75e41f5edc : SUCCESS in 5m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5b72b581b47745f7bb7a93a16bccc761 : SUCCESS in 10m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/63648a518f55485193fdf3b7b68a166d : SUCCESS in 5m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3a3d0a439fc1421484a01d8fd88b7896 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d943c01a6b90495f80c15e7d31d44833 : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f481047e55774255848e3edf28d2d446 : SUCCESS in 8m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/763d3fd482404a6f8874e986ff70e24a : SUCCESS in 13m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1a9c9b39f62d4bcb8ea1e5a1586952c7 : SUCCESS in 10m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/98fbc05c380a402481d09ff5d69c1ba9 : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dab4d518a609414cbb01cb999a7769b2 : FAILURE in 26m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9bb87b0ef1dd40afbf256c9ab789c02b : SUCCESS in 11m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/42e5d8d091c744bbaeb7ff718dc4c64f : SUCCESS in 4m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8df87d18fad94e16a418dab1924d0b06 : SUCCESS in 16m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/25ea9087f3844c5f935a3742078b3740 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/842fa38f9a6345dabf0ebea268a96d09 : SUCCESS in 7m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/974ba6e83a014a08b51a10e252a7f4b1 : SUCCESS in 18m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41b49956f7b64f0b8b7665d015d9a7e0 : SUCCESS in 41m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff95967b0a164d7d866260a310881866 : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ac24eb2a7b4c3bcc9db1f261c7259c0cc948e636","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-23 13:01:34.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"662809d857323983e85118ae7965f8cb418cbf65","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-23 13:02:14.000000000","message":"Patch Set 2:\n\nhopefully with the neutron fix added this will be green","accounts_in_message":[],"_revision_number":2},{"id":"14a23eaa8d0fa23c46160365a21b75d616ec807b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-23 13:20:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2de089df389e11a4a1281d9cf5d83cd440792a9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 13:32:08.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5af763af3afd425f8b5692f4886ae6f2 : FAILURE in 21m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c69bcb8fe216671473df0aff96ce0516f8c1c7a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 14:23:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f496452935f46f39dbd78788e2f8afa : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0df814c4e83d41649470c5e5d1af7581 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6bd977ef08e454ba420debd66df610f : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeb6237480644db99fec68548d6dc4a7 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f2d0d8c9d6c48b5bbe81dd92d120c2f : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/691e2c05f7b54d4284f2699d4840b143 : SUCCESS in 5m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c63fa22ea7cb40188acc3e46e8222ae2 : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea09e29e28a641dbb4edfb633ec62ec5 : SUCCESS in 10m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6f137d0a6da54241a8415e4cd9d9a109 : SUCCESS in 10m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c86f92a746f544a39f0ac23f270f46bc : SUCCESS in 7m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ba2d7989389c4327bf015e6684aa09ae : SUCCESS in 10m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ecf99ed4741e4c36b28d539fd8a2d2e6 : SUCCESS in 6m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/debd31a1c6554af8a77a54ba1b9485d1 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b193a8d70ae4242a523575e0c770438 : SUCCESS in 9m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6485e7c4ba564482b86d7da05799e990 : SUCCESS in 10m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c7ba6c2eea4b4379bf82f0e453e3e7ff : SUCCESS in 16m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/73dd42f1da8b44b2a2822745d30753d4 : SUCCESS in 11m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cca8431875f54567b85b96d667d1e24f : SUCCESS in 6m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eaa537e9b536403bb2d94d6bdc0fbbf4 : FAILURE in 29m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/421ee18ec1d8408c8801d7f6f45a48d0 : SUCCESS in 14m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7cd052fe968d406e99949a60b4f9b27a : SUCCESS in 6m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10d009f130da4d6c91f35239add29d19 : SUCCESS in 20m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0116d3a74b564ec39faa79f2b8496e93 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dda39203c5ca458aaa5532d5f0b87028 : SUCCESS in 8m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/30c6831496c84b72acf6e59e19f81f59 : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/649236895bbf40c0803038bfd5dfcede : SUCCESS in 47m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f618b79193584cf79796be7dad806e59 : SUCCESS in 1h 17m 23s","accounts_in_message":[],"_revision_number":2},{"id":"593b771c5fb4c625981c394072173fd55b3a946a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-24 07:26:13.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nazuu","accounts_in_message":[],"_revision_number":2},{"id":"7c8defb41f3eb8bf16db9745b1c2b0479ebb1845","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-24 12:57:27.000000000","message":"Patch Set 2:\n\nrecheck dependency updated","accounts_in_message":[],"_revision_number":2},{"id":"7ea8005785c29398c32e5c9af2d03051b18d584c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 13:21:58.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dae6eda340a34fdb81a594ce7fd9420b : FAILURE in 18m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5c3620a0182f9c5648aa1a91ba69907aa5c75765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:22:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dca385c68ca474faf1b9a1c47aad116 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5450d9f22bbf4485bbd769fb834f3a60 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/125149e5a2b945b4bc159a1e7d184a44 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/add67fe9865040ab9f9d7aa16949e04b : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/988c6a8e993c4abb98d57d88e83e466d : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0268c66e0a78495dbcdc13373983adff : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5dd1b5f034ff4c70a37b760ea4be6c8b : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e4c37205ab624988aaa25e7daadff6a3 : SUCCESS in 13m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a12c5e0408e04d69a9ce38f6205a604e : SUCCESS in 10m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8a633038f27a45bd9a8c4cb0fb257bdb : SUCCESS in 6m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f95717978a894a4aa764eae709f74752 : SUCCESS in 11m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4a41e856a4874ab2a5f308a7009df641 : SUCCESS in 6m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/65f73c1012854cab97cce1ca6b0eb71f : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/914c35c5817e4bc4a62d12ec12026d30 : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ddd5eeb01ce14da4b63274e04a003db5 : SUCCESS in 9m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6e4a5d0f6f42408090be48f9540c24cb : SUCCESS in 14m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5678dd5ff32049fe954022a7b570aea5 : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b0b34ba7d9e444e78829d3ab81bd08bb : SUCCESS in 6m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f964a9ace533476cbc0f762dcdbd80f5 : SUCCESS in 24m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9c723f24cf5840a7a061782b3bb78082 : SUCCESS in 13m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/07e6cfce46f64a1f95049540e0ee8716 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf954a0428734c16b071b79f7b906d3b : SUCCESS in 18m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/22dd1f6549084043b64c7cbe41236f7e : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/041d25a2e44d47e9bf163f8c82b960db : SUCCESS in 8m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e15ca24887f641a5954cd20621903abd : SUCCESS in 17m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51d6fd1b5cb3423b97a6220828d715e9 : SUCCESS in 46m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99983a406599499c91b1b3d4018afc56 : SUCCESS in 1h 23m 04s","accounts_in_message":[],"_revision_number":2},{"id":"81bbcd49dd686e5d43540e43884ce3fc22c9af08","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-24 14:25:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb01864450a13ffea4cdb60e42fa596c1010af5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 07:12:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e7c8f4e29f583f4599c70dcf9ae8968d8b16c13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 10:16:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d57a1dd90fc24bb1b4df724a14207488 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fecdaf2298b4e3d873a3cfb4af5669e : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81895880e7444f14b1e75b1a21192386 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93d1203caca84262b77a94d126f7e804 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a436d25d530417da1906920be4c473e : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7ebc3765b25497b8a4428b4b8afe331 : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b2dee2564ab4791ad1b0cda756d0a56 : SUCCESS in 10m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c53eb314196f4e52ae4d6fccf2c35188 : SUCCESS in 10m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/da2334d2615f4be5a9b558d62aceb685 : SUCCESS in 9m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0368fa3b98f249a0bb2e225b86614ad5 : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/324c265db04f4200a848a6a08ea4ff87 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54eb4b0a047d4db0b8efb87368361409 : SUCCESS in 10m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c97fe9b507d8450c8396a68704146c29 : SUCCESS in 8m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/16910cf9cd77458cb534e6f76e612e50 : SUCCESS in 13m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/60c4554846954f238232065346218ba9 : SUCCESS in 12m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2461bf24462c4958a67de53e90c7b919 : SUCCESS in 7m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7422b127fb63493ea8317f361df881a1 : SUCCESS in 21m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ce5e0c2cb65a40929d2760b1b8154924 : SUCCESS in 15m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c4d35f37de5f4b75b3d029c592209c5b : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/47557310514e4167bac5acc2ed234144 : SUCCESS in 18m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/39eea960c5bd49ffa6a49b61465998e0 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91742bf374924d11a0a1d02c163a6d6b : SUCCESS in 8m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0851dda54f51427e8ef797ea71f8427e : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f02d4321e7224bc9bf35812b617dbaca : SUCCESS in 45m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa7c63913d4e480284f1955d3eec3971 : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a9dfe4b8cda9bd4ef1c4aa1806449a0cfe2f9363","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 10:16:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fff7f786dcbe33ebe9708ca3c6836c4935cb3ab0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 10:17:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aef5cf329fa4fe99f760b07fb7be225 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d021d02eadb3475bde66335929f53f3c85df80f","revisions":{"16330bb2b51579bb07d80140988f76d36818a11b":{"kind":"REWORK","_number":1,"created":"2021-08-20 08:18:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/805352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/805352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/805352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/805352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/805352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/805352/1"}}},"commit":{"parents":[{"commit":"6562d0824f15990c01e2c1f8067c1853b595a534","subject":"Merge \"update constraint for monasca-statsd to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6562d0824f15990c01e2c1f8067c1853b595a534"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-20 08:18:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-20 08:18:31.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.14.0","message":"update constraint for neutron-lib to new release 2.14.0\n\nmeta: version: 2.14.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Co-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Change-Id: I9a6245e4b38bd329696e3c6277e33bbd33a0d26d\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic8bf060742b360d5905ec3d93a386b9f6bfc2600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16330bb2b51579bb07d80140988f76d36818a11b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16330bb2b51579bb07d80140988f76d36818a11b"}]},"branch":"refs/heads/master"},"2d021d02eadb3475bde66335929f53f3c85df80f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-23 13:01:34.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/52/805352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/805352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/805352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/805352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/805352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/805352/2"}}},"commit":{"parents":[{"commit":"6562d0824f15990c01e2c1f8067c1853b595a534","subject":"Merge \"update constraint for monasca-statsd to new release 2.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6562d0824f15990c01e2c1f8067c1853b595a534"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-20 08:18:31.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-23 13:01:34.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.14.0","message":"update constraint for neutron-lib to new release 2.14.0\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/805627\n\nmeta: version: 2.14.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Co-Authored-By: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Change-Id: I9a6245e4b38bd329696e3c6277e33bbd33a0d26d\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic8bf060742b360d5905ec3d93a386b9f6bfc2600\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d021d02eadb3475bde66335929f53f3c85df80f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d021d02eadb3475bde66335929f53f3c85df80f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
