)]}'
{"id":"openstack%2Frequirements~805474","triplet_id":"openstack%2Frequirements~master~I4cee198cb4262aa2d7b2c2b519bbf830942d261c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I4cee198cb4262aa2d7b2c2b519bbf830942d261c","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-08-21 06:24:51.000000000","updated":"2021-08-21 19:08:32.000000000","submitted":"2021-08-21 19:07:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805474-openstack/requirements/constraints/noclob","meta_rev_id":"9cc65c1547935e619051b30f911e622642199df0","_number":805474,"virtual_id_number":805474,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-21 19:07:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-21 17:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-21 17:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-21 06:39:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-21 08:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-21 16:15:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e9d36dbd152c55bfc442ca357d92f0d1bac8c51b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-21 06:24:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92af6bb00bc3b5a881203042144ae2b5be83b638","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 06:39:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4baef7a1a9284b24bd2eaaecc6918b8f : FAILURE in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3bcb24279bddb5834c478e0734dc0b08ed13d0c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 08:00:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6599dfd30cec45e089b704072b732c7e : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fbee0b91a7345a2917be77744d3cbbe : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fc27e538bc240449c3a61b296989816 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c548aba2d65e4490bae3996d53cc5347 : SUCCESS in 9m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4dbc6d0ced044d1b707dd3130dc400c : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be612ba5455a4b709ac87169663abdd0 : FAILURE in 5m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5916ae58f53f483c8507dc7c48ae2593 : SUCCESS in 7m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/273bd6a862d547ee90bac7dbda6750af : FAILURE in 7m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/987d455144e3438c93f6bad46103866f : FAILURE in 7m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a178c2b8883b4a9dadc31e960189feef : SUCCESS in 8m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5e98e80d4b6d469485bb29d0853012e4 : SUCCESS in 13m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/10450bb2788d4ba6a88e7ff1b376ac15 : SUCCESS in 9m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/806c304d745748e582b7caf378b3937b : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42c7b852dfe7442f965a86daf45cfd5b : FAILURE in 9m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5e93d47d385c491d8a750a70978b4eaa : SUCCESS in 10m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/80d17ec92ce14cdab54e14676479a67f : SUCCESS in 14m 46s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/27ecce82b22342cabad647b461d5e01d : SUCCESS in 10m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a4babd986bb1467a866148a50742598f : SUCCESS in 6m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/280eb1986256444e890f12d4991f1548 : FAILURE in 25m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/37cb9b95e55f46b79395d4758e358af8 : FAILURE in 13m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e8c231a6da1c472ebf67657c1cd9d7e1 : SUCCESS in 5m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/511a870b29594b6bacb0a26fcf5c0fee : SUCCESS in 18m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4cc3f03d36204128bc000c56571846f2 : SUCCESS in 8m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2777ff2a1ca94757af577f88975658b2 : SUCCESS in 9m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/690399bfda064bce8f9fff88eb365490 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/146aa5262e354ab4a1f79bdab3f1232e : SUCCESS in 46m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed7cb0c14e5c49959a0a42cc8c3314ba : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e2d3577feddec7e5b4441eef7c4e5ce7157f9b65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-21 16:15:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64744fcf6cb81e208c171c8271f9c2e311e5670f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 16:35:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/abfc5ae9b9624af7901de8a56932b95e : FAILURE in 17m 39s","accounts_in_message":[],"_revision_number":2},{"id":"658ef85b57e51a01356a6e40a5fd2ee6a9cd5251","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-21 17:02:10.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"12a757a472e476f9bef37221d65ad927da7e9fca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-21 17:02:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0251e78180669f089d6bdd20cbac7f2116c45ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 17:46:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98e131803a894cb38cc1ae9bb884246c : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c1248562674afba8ca72f2b5517d84 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/077a4e132c7f43c8b9174636adf65824 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c68f84fcc69940bd8f65dd9484cfcaf2 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbe0678130a241559192c1a871cf9881 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e90e81dbef6443f2963eb47f1423f5bb : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/788902be6ad24e85a2dc096cb1f74d75 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/14cb87bd2e994071a7d96505b0fafea3 : SUCCESS in 10m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2711686407e34d3ba1ada8b9b33df59b : SUCCESS in 10m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/eb756c1b83db4a66a0483946057263d7 : SUCCESS in 5m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d1bec36bb5844eb5b97c801b74cf07cf : SUCCESS in 9m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/59fc11d0b78f4c3ea62297816adfc917 : SUCCESS in 5m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0b67ae821f654048a15d4bb81509616b : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8e8689e7946b4c99929fe8b4d5338a9b : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/10e221f743cf4501b8ae353173d07277 : SUCCESS in 8m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/90526fbaeaf645749f519dc83ddee15d : SUCCESS in 19m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/613c96631493470a88ba0cf50cc9fa1f : SUCCESS in 10m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d730d2d9b31641d0806b1f6231651ffb : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5d79d354d2ca4eb78e02366c017b583f : SUCCESS in 20m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1e09c8d0e1cb461ea88c62820eea0ac3 : SUCCESS in 11m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/33b4e35708ed4d038addb835c8eedc98 : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e90969199e2498c9775b67facbc57b5 : SUCCESS in 15m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1b9bea78e6f94608b5e72ffbab2293a2 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fa286ccad8d44faa81dadc46c4d7728 : SUCCESS in 6m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/53d67e1f91b24cc59fa6eb6d734e3d05 : SUCCESS in 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a814a0bdf2b74f8bb6482d715b25725a : SUCCESS in 42m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec393c66a8e043faa627aa4f975ea3b4 : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":2},{"id":"d9d2afadc0b511354028e7d58a4f977bb6b8b827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 17:46:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1b6b81a4e8b4adc81bbd3cdefacc5071cff8ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 19:07:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8c000d90f04630bd1858c7492ddcfd : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda440c928f74ddca2c38d0c9178387a : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3100e5a2ef2c4254855a893a67d5789f : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4601e805d7d40159ea4e37f7425d56a : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/624872bbe7254e29bc5d3b1dee64894a : SUCCESS in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04290a4d0bbb4dfe8ac16eecc4f1b35e : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f289642a77347e884441bc78a8e95aa : SUCCESS in 13m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce21a53fd70b48158e04443517c3d507 : SUCCESS in 12m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f37b6b99718f4a6ab1e06930535cba93 : SUCCESS in 12m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ddc3784dc3c24c73b10487dff8132b89 : SUCCESS in 5m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d1276af3aaa445ca82cfce8d90d1a0b5 : SUCCESS in 7m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b3dadcb839b4d68870d18b4691a595d : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d27b755102ab40709e2cacb30a4c03b7 : SUCCESS in 6m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5d19d9281779433c83d75c16f90929c1 : SUCCESS in 11m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e3b809ff90674fd8a063ceecde1e8bb5 : SUCCESS in 8m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/667d6277f1af4afc89838f6a21299640 : SUCCESS in 7m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1f1f90f1cea14228a4b73ede5927f67e : SUCCESS in 19m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/340ca64699e246f1b348cab1dd5605e1 : SUCCESS in 15m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c49f0b04f6a4b3a840562b6c05e7f41 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4329518630f945138d5b1822efb1a53c : SUCCESS in 15m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/880064f7747d46fd8ea2bbf3ebb20e2e : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bcb9d47eaa6542c4930485c83fdead65 : SUCCESS in 6m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9944ee7d9c594dc4bcf1a566d153fe68 : SUCCESS in 16m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7c409753c2441d1a9b1cbcbb7af7bb6 : SUCCESS in 45m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9503181bf834766992e67f5dbdc8ad3 : SUCCESS in 1h 17m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d976ddaac4a21cc905534c345d22a9b61603d850","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 19:07:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9cc65c1547935e619051b30f911e622642199df0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 19:08:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba338c1063f4f2c9e75588029ddeaa3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4bc55806a987e0b3257936294a18a5409c4f6b48","revisions":{"6c88dcad33e2bbc61141ea7eb620dd532959be3a":{"kind":"REWORK","_number":1,"created":"2021-08-21 06:24:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/805474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/805474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/805474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/805474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/805474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/805474/1"}}},"commit":{"parents":[{"commit":"002bb1a879b4d068bc42ae1964aa694a109fd7e4","subject":"Merge \"update constraint for heat-translator to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/002bb1a879b4d068bc42ae1964aa694a109fd7e4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-21 06:24:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-21 06:24:49.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4cee198cb4262aa2d7b2c2b519bbf830942d261c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c88dcad33e2bbc61141ea7eb620dd532959be3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c88dcad33e2bbc61141ea7eb620dd532959be3a"}]},"branch":"refs/heads/master"},"4bc55806a987e0b3257936294a18a5409c4f6b48":{"kind":"REWORK","_number":2,"created":"2021-08-21 16:15:40.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/74/805474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/805474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/805474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/805474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/805474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/805474/2"}}},"commit":{"parents":[{"commit":"002bb1a879b4d068bc42ae1964aa694a109fd7e4","subject":"Merge \"update constraint for heat-translator to new release 2.4.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/002bb1a879b4d068bc42ae1964aa694a109fd7e4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-21 06:24:49.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-08-21 16:15:28.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4cee198cb4262aa2d7b2c2b519bbf830942d261c\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bc55806a987e0b3257936294a18a5409c4f6b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bc55806a987e0b3257936294a18a5409c4f6b48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
