)]}'
{"id":"openstack%2Frequirements~805501","triplet_id":"openstack%2Frequirements~master~I9826c22e82e1f5f0957c2bc4e5baf4bf224a7739","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I9826c22e82e1f5f0957c2bc4e5baf4bf224a7739","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-08-22 06:36:37.000000000","updated":"2021-08-24 20:39:53.000000000","submitted":"2021-08-24 20:38:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805501-openstack/requirements/constraints/noclob","meta_rev_id":"d848b3deb0d206cdd9fe1fdcdd5fa845faf011f4","_number":805501,"virtual_id_number":805501,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 20:38:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-24 17:25:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-24 17:25:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-22 06:58:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-22 07:57:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 16:09:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9926a32650312a0b168b1d3fa7eefed59952aadb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-22 06:36:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e48af4defbaa80b6ab579c10a608cc8c90b864d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 06:58:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/901e08941eae43dfa79f41def9d8ca2c : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bd5e54d03f36ba2c64fd7e665ead081aa102d177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 07:57:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c5939c6c284ce5b3fe3b7fd5c3bdb9 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d63d8d5c7f4b7d9f5fbfcf2f640f32 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7f6728b5a2b4ed6b498b5ad93cafff1 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8afa44db722b43d5a29c8dbc7876d97a : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e050d2b171f4e6a9c33ac01b32cfea5 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1622592004fd4dd78574d903a3736fc5 : FAILURE in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5bf55d6318f41ff9cd86665b70809df : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72e298d312c7492b9380b72cedbbd6f4 : FAILURE in 6m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0971bd3b01a04ac7a3ef0f4dcb23b76e : FAILURE in 4m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/277f11025b24464f9b1fd318bd31502f : SUCCESS in 6m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d6ba5b71ab144438b3f86096ebc77255 : SUCCESS in 9m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c5c4ffd1ea9f48b487a8729f9cf8b46b : SUCCESS in 8m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/33d015c0145d47d0bdf1a780a6b72aca : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff044c32277c4d3eac87d997a25f0520 : FAILURE in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/01852e5a77db4996908f3e1bb43ff265 : SUCCESS in 10m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a80cd4616c2046e8972b2d59cc3105db : SUCCESS in 23m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d8af3817a0b24141bd792008d6bf9af1 : SUCCESS in 13m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fd4ee533c01e4777b3556594b8b058a8 : SUCCESS in 7m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/04b9dc3d2e024130abfc88e8b270be67 : FAILURE in 27m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/58dd457753fc447e8818e41dc0eabcf1 : FAILURE in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3ccac67be9864225a7f978594771b265 : SUCCESS in 6m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b80aa8e7d77b4b508b163603826dd063 : FAILURE in 17m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7e68f907854c4c44801cccaf25e2dd99 : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc5f7762bb9a40a7b26c191774615b48 : SUCCESS in 8m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b459a357942b4dd480d3eaf904f22630 : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c64acc716254664a0e951c368f48c4c : SUCCESS in 44m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aaca007d76c14f4fa38711f704efe9c5 : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"460ba7a834efb490c0df131b7e0a3fc8281b586e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-24 06:25:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec3b80d61b1bbaa120c9a9f8f74e61b49173eff6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 06:39:31.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d4b8a21bbc6415f9fc2702d52be34b0 : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"649cd1c9e2ac772d336e2d3633786a1dda34b1d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 08:34:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ddf9ade3874f479f6af117b57d2457 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d913364c66cd4dcfbee3311fedf916d2 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fb33aac4add444db0653177e48109ea : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef44037ad0db4c91b0775d5bb50bb40f : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6070d6e6161145ad8c6189de9cef08a5 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9283675faf5743ffb6794afed4dced0b : FAILURE in 6m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1932bec19b674f47aba92e3d0c7c2b7e : SUCCESS in 5m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b634fcdc07dd4935a5e6d27b20fc2ae9 : FAILURE in 5m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c2765bc434584f0bba1c5dd439f336d9 : FAILURE in 6m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/08c326cd32c74b11bb25b43cef010c2c : SUCCESS in 6m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6881d0be09ed44118eae86fb455c92b5 : SUCCESS in 10m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b57e53cff5854af3877e49a3825266cc : SUCCESS in 7m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b80f04d11563400187b88e83c22a3941 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03f3241566444edd809b566004ea7b62 : FAILURE in 10m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/93dc6d09341c417bb868d30b364e6ec0 : FAILURE in 10m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/321ed90c1687475995af52b8a2250388 : SUCCESS in 15m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ef1d9313208643bfbb260b5cbbdef766 : SUCCESS in 11m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5fcfaa787bca4881a5e225910af0a8eb : SUCCESS in 6m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/48cef163fba843579b959834bbf2059a : FAILURE in 24m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/598b3e229e8049589bf3b85078cc7b88 : FAILURE in 14m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e71937a891dd4369b8796ea8ed6df7ca : SUCCESS in 5m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c294269cb55248bb950152e34ae609a4 : FAILURE in 19m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7b263a8fb17f46bdaa20e190260d8020 : SUCCESS in 8m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/157a38b73704470b8c582082561801ca : SUCCESS in 8m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/97ffd8f0a3224835b5ba4722334402ef : SUCCESS in 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db1f8942ab1046149f845a9575284108 : SUCCESS in 51m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ba6fa746e274653beeb05398784aef4 : SUCCESS in 1h 22m 30s","accounts_in_message":[],"_revision_number":2},{"id":"dc60d5bce45d0b37776f62c6f24ec8037e6c4ad2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-24 16:07:25.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"0d13a6712614453816b4bb9d966e1538f17eba98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-24 16:09:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73dbb1af34f51be782f88ba5c5b055ea5b734829","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:34:11.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a0c03f83ea9b4822b49935e59063fc3b : FAILURE in 15m 22s","accounts_in_message":[],"_revision_number":3},{"id":"2f939e626a92355ff903def430853338229ab335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:25:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b38a7b18d1243cf908a4fd63212ab35 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd01298511c47c0b3bda84cc6081a31 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c55e68562da438f8bc0f27c036456d8 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c651533b928043188e6a45c77c210a71 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c10b3438281c475aac0d5214d88b5ef1 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ffff80ee52554348a076f2b9b21e09fd : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff7d11ef18254f8cb05229859c4c9198 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb08d94da1ac4f34bac47da9ea5f080d : SUCCESS in 11m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9cd7c342e48a45a38a5e18ed98a19940 : SUCCESS in 9m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4e8acc3a32404adca414b00078aaae55 : SUCCESS in 5m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a27bd3297686454a8369f9bd449fbcf1 : SUCCESS in 9m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9a57fdd9ecb84d7b8ee6bfbb9ea6bea4 : SUCCESS in 6m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f55f2b1dfe824b8c9933d316440262f9 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64af189886914a778cda3cde572cd0c3 : SUCCESS in 11m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/589f10c2843a4b069d82ac0b3ec1be35 : SUCCESS in 8m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fe4608e59daa4807a2f9393326379e05 : SUCCESS in 19m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/54800dd6ef7045ba8d0e75c1c2eb5cf8 : SUCCESS in 14m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6d34569850d54940b4b0b7e3166845a5 : SUCCESS in 6m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a6e8c82d00714bf4b09b4cb9fe569b20 : SUCCESS in 33m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3ca2ffde39fb4550bd334078560cdba9 : SUCCESS in 15m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eda30248054341c78f987e64163a85ee : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42715d2655404071a39674d119d64675 : SUCCESS in 18m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/89d94997e4f4467b96f2704744531812 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c174e95c7d574d56929f0d92c1bf3cdc : SUCCESS in 7m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cea87689ee0c4d698ba1812256c000df : SUCCESS in 15m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2f3c8977ab54deb8e0812fca90186e8 : SUCCESS in 53m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b0bca2054254102a950cf2c7f6a50fb : SUCCESS in 1h 14m 23s","accounts_in_message":[],"_revision_number":3},{"id":"db66729866319a4148bc8e0562fd2fd9d4cf5751","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-24 17:25:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"26ead100d3676e0326a14f36afc8d5d735b5e3ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:26:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3cc767a6d8ef25e5ac651fdb912da2f3811527a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:38:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e462c6d01dc04d9b899bc2fca03c3bdc : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b62684272c49149fd16791ef511255 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2019bb2c4375471fb1df895c1b90137f : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd93436e1b7a45c6a0e703e877349ae6 : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db60e8852f5f44f7a68d4bcc432b9f62 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3d155e89c424d4f934841487b356906 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/88145491a06c4b519f3fb807bca908d1 : SUCCESS in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7826d0889a94b0ba8938a440996f580 : SUCCESS in 10m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/13ab74ac49274c4c8880c512825b98d0 : SUCCESS in 9m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a0befe2e13b344079276881a9738f663 : SUCCESS in 6m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a92e3ecf48f743b59dde1bbcac0db7d5 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20f2ac65a81f45eea52bd7e1d13eb115 : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0d0fc6c4de3b46d0a0dd3a1ab9f513a9 : SUCCESS in 9m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/adbd94f688d346d8894a06ee6ae1ee38 : SUCCESS in 17m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5c9e3d1197b340fdbce7aebe4505bf74 : SUCCESS in 11m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/78429859bea4475e8db80fc8121f41bd : SUCCESS in 6m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/17a9cc9755b4463aa32a0a23b9a61909 : SUCCESS in 26m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3113eaa9bb6e4faf946890a6f7fb9e8a : SUCCESS in 15m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d246822289fe4c1582247afd82c64275 : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa7f551a0bd2424a8dd009644105190b : SUCCESS in 18m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/352fc29165f54e14bfaf173739ca5e7e : SUCCESS in 6m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23fbc64cf18f4e00864111cebf8158b0 : SUCCESS in 8m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f127d785cb34482c9fa82fe9b73d8380 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/866ec49fe5b141bdb2d7f73e2a960ffa : SUCCESS in 46m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f9b88eed3e949bda4b9ca6319d4e984 : SUCCESS in 1h 12m 56s","accounts_in_message":[],"_revision_number":3},{"id":"b417f29f7f440995b9c1fa954b29be183a6b0bea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:38:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d848b3deb0d206cdd9fe1fdcdd5fa845faf011f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:39:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a0d52407b84a6089ae497804b9f82e : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"26d7e48ee0e585eab06043894d8c202123ace701","revisions":{"55d710d3658b1a4566d2f8da65f9c8650c171ec5":{"kind":"REWORK","_number":1,"created":"2021-08-22 06:36:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/805501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/805501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/805501/1"}}},"commit":{"parents":[{"commit":"4bc55806a987e0b3257936294a18a5409c4f6b48","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bc55806a987e0b3257936294a18a5409c4f6b48"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-22 06:36:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-22 06:36:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9826c22e82e1f5f0957c2bc4e5baf4bf224a7739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55d710d3658b1a4566d2f8da65f9c8650c171ec5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55d710d3658b1a4566d2f8da65f9c8650c171ec5"}]},"branch":"refs/heads/master"},"5869163bc693304dd8c799c0faefaa6a15b23ee7":{"kind":"REWORK","_number":2,"created":"2021-08-24 06:25:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/805501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/805501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/805501/2"}}},"commit":{"parents":[{"commit":"5f97046379c3978a65a8e1ca60164d771c2847b0","subject":"update constraint for os-traits to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f97046379c3978a65a8e1ca60164d771c2847b0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-24 06:25:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-24 06:25:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9826c22e82e1f5f0957c2bc4e5baf4bf224a7739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5869163bc693304dd8c799c0faefaa6a15b23ee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5869163bc693304dd8c799c0faefaa6a15b23ee7"}]},"branch":"refs/heads/master"},"26d7e48ee0e585eab06043894d8c202123ace701":{"kind":"REWORK","_number":3,"created":"2021-08-24 16:09:00.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/01/805501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/805501/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/805501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/805501/3"}}},"commit":{"parents":[{"commit":"258e31533c6ac0a80565a65fabdd8b646698541f","subject":"Merge \"update constraint for oslo.messaging to new release 12.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/258e31533c6ac0a80565a65fabdd8b646698541f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-24 06:25:33.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-08-24 16:06:42.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9826c22e82e1f5f0957c2bc4e5baf4bf224a7739\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26d7e48ee0e585eab06043894d8c202123ace701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26d7e48ee0e585eab06043894d8c202123ace701"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
