)]}'
{"id":"openstack%2Frequirements~805612","triplet_id":"openstack%2Frequirements~master~Ibb14983265f510ca09bb4f76c1d807e787b6b4a4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibb14983265f510ca09bb4f76c1d807e787b6b4a4","subject":"update constraint for os-traits to new release 2.6.0","status":"MERGED","created":"2021-08-23 10:41:43.000000000","updated":"2021-08-23 19:34:21.000000000","submitted":"2021-08-23 19:32:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805612-new-release","meta_rev_id":"571dbdd441b60f4ccf852df1c98510f573454539","_number":805612,"virtual_id_number":805612,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 19:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-08-23 11:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-08-23 10:43:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2021-08-23 11:09:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-08-23 11:47:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-23 10:43:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-23 11:03:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-23 11:09:36.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-08-23 11:47:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-08-23 11:58:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af5839c3b4a5567acbc4282dd847a83f4445573c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-23 10:41:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e0567d9a2907fade7097a2521798a1c956704be","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-23 10:43:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3c738ac621556de2835238a8d2b7bba005dc1e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 11:03:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/76f6e2a8986847e3b9e16fd863cc5294 : FAILURE in 17m 15s","accounts_in_message":[],"_revision_number":1},{"id":"630f0c0f67cf5f1d9e2e71047243e9021cc6a13c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-23 11:09:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"be3afdd575f026a56f60a4adeb0a7e701f5aee40","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-23 11:47:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82e2003df61bfb0288fff8766d9ebb2c70cf145f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 11:58:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72bf640935e48b888a840a75764f01e : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb5f772beb24bbe8a4ef7f7eeecd03e : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa74403151514b059297b341f5ac965b : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44c3bad5987b47088d570bb2d157ac81 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/928aa163c44e4fd88d2fccfc16bb5eea : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5bb0691a09174b79a1922b025c270dfc : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d62037fd9d8e41d6a24ecb987d54d6ee : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/42f9593a979b40fdb5f5ee41d206dfb1 : SUCCESS in 10m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8a0cbc11a22c470f827da4f5341155e1 : SUCCESS in 9m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/82e4f5808196444e842cf6db573824ae : SUCCESS in 7m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4d5d2630bbf544e2ac49d0119f23ae70 : SUCCESS in 9m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/56f03c3fa6d74a0a875b9189f925fef6 : SUCCESS in 6m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/84f0fbb7d28c4a7e99c9a81c05944a50 : SUCCESS in 8m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0565e4f63d144a3d9dd73e306b7e55d9 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/433e74e9c5b543e5b35e91674c029478 : SUCCESS in 7m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/449926084eb7443b9ea7fea1eeec8cd9 : SUCCESS in 17m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e10ea92163e5424bb23a51c2c06d4ed5 : SUCCESS in 9m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6518dd99a8304f2b9b3cd95fc30df7bc : SUCCESS in 6m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/332fe2ee223f4889abf6ee8744c015d0 : SUCCESS in 25m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/140d5807267e4238b647e7e0312d75cf : SUCCESS in 11m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f55b2e8d66e04cd6a19a65c88186f9bd : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7384b8b732de4a499bb5d3110daa4b03 : SUCCESS in 17m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a503cf0a05d8400f851a39ed802d01ad : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4fa198fa7a048e89f0c643493dcbc4b : SUCCESS in 7m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fc2195fd0d014797b6ff44bbf6f5a76e : SUCCESS in 15m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba64c75487c149688844ad83fccb2129 : SUCCESS in 46m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79b904cb26f141e2a9e78f945424ea3b : SUCCESS in 1h 15m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fc622af49804b741d0eb96f679340d3fca06d53d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 11:59:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e9a0cf6fad2e838646102462f65a02f16308235f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 19:32:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edc4707d47549e8a4bbc19ab054615b : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7318890c7ef64d0a8fa9f29c3eea1886 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30c8b3f9e54e46069b8d33b7d0c6096a : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccd9231fbb1041639080f5a48ab0ba89 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6807ffb498b34ac59391a47e70cef1a6 : SUCCESS in 5m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a78b643a641548868b74a5944f43afce : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af345df716c249a08335caaba7a2fce1 : SUCCESS in 11m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/873622ba50c84f98b70cb3b2a48c3660 : SUCCESS in 11m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b48bcfdd155e4a22bc1a191577788e18 : SUCCESS in 10m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a7f391f4a57042f5ade0edd5c45e7055 : SUCCESS in 7m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cd85b950ce304675a5898522dc6453b8 : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9be4e6a152334411bcb311f13b246d30 : SUCCESS in 9m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e29e6155037f43048ab56ad277fff3f4 : SUCCESS in 10m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cd3bb43c82084f94b3c9bf98649005a7 : SUCCESS in 15m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/29502735c2af4900a1631693a42cdba4 : SUCCESS in 11m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cecc17900c48482fa6d22c2cc6b4422e : SUCCESS in 8m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c2b4ac24664445f2bf7611b4d5b4386e : SUCCESS in 26m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c04a508ca6b14276924dff36ec452b74 : SUCCESS in 14m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95bd919520944535a07cdb4b4e9b866b : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee95000b996f43428eba9e911ecb293c : SUCCESS in 19m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4aa429ed22004736b6bbfab8a8453bbb : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bfbada0355e04f0a9d94d35f54737c8f : SUCCESS in 8m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/16178a30e2724122a7d0a0dbee3b5d5a : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1d4b08e9897498eb25faf156aeda196 : SUCCESS in 46m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd646b24500f4cfda19eb3308333645b : SUCCESS in 1h 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c11ce6a08e1205544de8db4d30ce666c58ccb008","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 19:32:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"571dbdd441b60f4ccf852df1c98510f573454539","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 19:34:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e557497e6a94ad4821ef79230eb9db3 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f97046379c3978a65a8e1ca60164d771c2847b0","revisions":{"5f97046379c3978a65a8e1ca60164d771c2847b0":{"kind":"REWORK","_number":1,"created":"2021-08-23 10:41:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/805612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/805612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/805612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/805612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/805612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/805612/1"}}},"commit":{"parents":[{"commit":"4bc55806a987e0b3257936294a18a5409c4f6b48","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bc55806a987e0b3257936294a18a5409c4f6b48"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-23 10:41:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-23 10:41:40.000000000","tz":0},"subject":"update constraint for os-traits to new release 2.6.0","message":"update constraint for os-traits to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib7ea3283180820bac345802912488ec508b78d47\nmeta: release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ibb14983265f510ca09bb4f76c1d807e787b6b4a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f97046379c3978a65a8e1ca60164d771c2847b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f97046379c3978a65a8e1ca60164d771c2847b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
