)]}'
{"id":"openstack%2Frequirements~805710","triplet_id":"openstack%2Frequirements~master~I820f80b0c423a952f6625375c7ded50296476b49","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I820f80b0c423a952f6625375c7ded50296476b49","subject":"update constraint for oslo.messaging to new release 12.9.0","status":"MERGED","created":"2021-08-23 21:10:53.000000000","updated":"2021-08-24 15:50:14.000000000","submitted":"2021-08-24 15:48:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805710-new-release","meta_rev_id":"91a6c418b58014a6729ec288e70c03ee0ff78693","_number":805710,"virtual_id_number":805710,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 15:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-08-23 21:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-08-24 14:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-08-24 14:25:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-23 21:11:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-08-23 21:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-23 22:54:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 03:15:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-08-24 14:25:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5774dfd9c61d44d4088f4c8009740b60173167fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-23 21:10:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8d3ebc1c1dd3c2239a94eee9c93cf6a456e59b0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-23 21:11:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ae3d98b95a425b285cd7de42f90d4fb338651dbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 21:29:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c2625defb3e4453810186763b879609 : FAILURE in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c1c7e60ebe353e3b4fc3fe18ef0eb62da005b08f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 22:54:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6dfcc5b70d4916bdc20a2ea733b235 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91204d8e1d3b4b35b77e6fea0b8ed517 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1010a89f3c04f61abcd27f929813758 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1f0977bf9d244aaa4914f538101f246 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8bda38955e04fd09751f72825e651b1 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/803cccd090c34bdab7ac135984ba2400 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5b614f5f0104917adf36025e8fee41f : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc4eaa9ab0cc402a9f0ddb4aea899bee : SUCCESS in 13m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/07ea55c86e4e44edb73a1c204ce41da9 : SUCCESS in 9m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/967f8ecd54d642ba8eeea44ae2219a79 : SUCCESS in 5m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/34c6c71b3a994b179c0fae0c324869ed : SUCCESS in 8m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c38434a94c5a44b49c60ca2292e44600 : SUCCESS in 7m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1e7d9d94fc694b018f8d58cd17081c3f : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ab8b05708354d0a857c86cf2af7f446 : SUCCESS in 8m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f4a8685cfff940148712bfa17d8dd8a8 : SUCCESS in 8m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cd9d295d357d4e559ae7967a8afdacb5 : SUCCESS in 19m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/61706ae6dfc74c41a4e1d1d9fbe34467 : SUCCESS in 8m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3bd07a48828349aabfeceb275fc08756 : SUCCESS in 5m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/294f7658db1e4a6b84cf216d870eee13 : SUCCESS in 24m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a317f27ae1284c05ad9c8284aa8b3b78 : SUCCESS in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ccf64c1b9a3a4c9bab7cdec165a646fd : FAILURE in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/447f72e3803545c0a8a8e2247c36abc8 : SUCCESS in 17m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3d5c001b5792443bab2cb5f92cd09a18 : FAILURE in 3m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/185e4cd3179941c28d3e1029c436b0c7 : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/65c97ed9f31d43559e28a78a5917e123 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa9b387f7af94d51a24e2d39116449be : SUCCESS in 40m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9be14eb4a8a74185acb29ebdb7f5f659 : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6e46ec1422d1e8a8e9c159d0a4aa863ac70679f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-24 03:15:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8fff954f7c6237da780086bdb5bf5ab848cf2e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 03:31:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a958328699784557bfa81622950bed3f : FAILURE in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9c2a946c9da5e4b6d409ded626348389f63728db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 04:35:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f6630a7b15a488994769a0a2646a515 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/098dc7fb276c4722bf30953ee292d0e4 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c2a0033dbb14923b86fb45ac39a4e5e : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/417e5b474f834981bbe377b925ed62c2 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/599ba18b96224d94bb2a478261eca846 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8f96ad4473b46ac93d65567988fb50a : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21b1ab93580e459995341f392f9ef102 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c08e54022aa048f2b13e92debb7ef31d : SUCCESS in 10m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cae45c2791dd44569b7c3cb44a0bf147 : SUCCESS in 9m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/115cccfe501f4afb8a1813edce75efde : SUCCESS in 6m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a875e26c72c44cb097fb987af696e72f : SUCCESS in 9m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/467ae81dcd5e454a9cfca2c795033861 : SUCCESS in 5m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9c67cbac8edd4b1e859ad00eb03fa983 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29975c7966ed499e83ff06c4b9681de4 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6a06879f2ef44c469a0404213e5a41f6 : SUCCESS in 8m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/782d7a93a64b49888ed9c52abfe884d7 : SUCCESS in 16m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0931d6c41985429bbe3ae295e010e2da : SUCCESS in 9m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fd5825d0cb69498da7b393c2494d7c00 : SUCCESS in 5m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9ba97e03b2614004b9be44cb53d62b6a : SUCCESS in 23m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c4f9531667b44fb091058d574f9ebb89 : SUCCESS in 11m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/16a0623027c1400aaab1c544ef1d55ef : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d6a374b0b6894e729414477ab059957a : SUCCESS in 20m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2cf3c0ad34814021a73535d7eacbbae6 : SUCCESS in 5m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1471a3b833564d1e82bbb2c93dfea39d : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/943a4b9abff6414bbd9696242026f977 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12849e353e3c4bc8864825231c9f75ba : SUCCESS in 45m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b166177aa424656b73397852da5bc58 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"01b8e4c334047937ebb07160a4bf262b8c39c289","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-24 14:25:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fa9b908863baab72b2c779301cc991ff85c2937","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:26:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b0069c634ef73fda059c7d3a0e4fe0cc4beffc8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:48:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a768fa2fe764201939aa198b9cfc861 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b5918bff40f436f8010643be690f900 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1109310147944cfaa5a6670fc53c7350 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b5844d54dd4df9984513dab2521012 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/691bbfdae36f49469176ccb7d76dcf3d : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/80f8b3cefc6e41b4a34b68638c2aead7 : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/188eddf3d9724be0ac5eff998cf677e9 : SUCCESS in 11m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ddb2f1e986f04700b009e0347a591dad : SUCCESS in 10m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/aafa4810853e4981b054e9f4828ad88a : SUCCESS in 10m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/efb6628a2fda43acaf9807530872b48e : SUCCESS in 6m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cafff6e0cd014ba3aa048e7d5b63cf6e : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e35ff756be054315a7d520fe238c0289 : SUCCESS in 9m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7417cd1e9f7e45849c23fdb48584c874 : SUCCESS in 9m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2d2dd008868246b48f8a80d02781761e : SUCCESS in 14m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/16c98cc87cea45289ba352a6f671e9b1 : SUCCESS in 9m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/884a618233e64b34854cf7068da814d6 : SUCCESS in 6m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3631013fc085483da21d772a053ac1b4 : SUCCESS in 25m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/30b1b0a194ff48bc801482129759641f : SUCCESS in 15m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a14a76cb1be4361885ad367bc63eb60 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8f9e3546409403686816de176ed208e : SUCCESS in 18m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/92da964cc792454db0511856952070cf : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a44b3463851b402cb4f94329e4a4eff7 : SUCCESS in 8m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/50f7b09fe4da44eeabb3f7ef211e768c : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff55ad1a771b46df8db3982140c72218 : SUCCESS in 47m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e39f407b2e66409fad61c7b8ad1a4dff : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":1},{"id":"24218825200b7e88179c9b7f7cb58dd700d82dc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:48:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"91a6c418b58014a6729ec288e70c03ee0ff78693","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:50:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59c54d4eac34b468ebeab5115acbc8d : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d66f48d27fda931958693c42780d070c44f4ff5b","revisions":{"d66f48d27fda931958693c42780d070c44f4ff5b":{"kind":"REWORK","_number":1,"created":"2021-08-23 21:10:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/805710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/805710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/805710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/805710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/805710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/805710/1"}}},"commit":{"parents":[{"commit":"5f97046379c3978a65a8e1ca60164d771c2847b0","subject":"update constraint for os-traits to new release 2.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5f97046379c3978a65a8e1ca60164d771c2847b0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-23 21:10:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-23 21:10:51.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.9.0","message":"update constraint for oslo.messaging to new release 12.9.0\n\nmeta: version: 12.9.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: If894ff6d4cddcc7aec3ef9254113469e2c92b099\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I820f80b0c423a952f6625375c7ded50296476b49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d66f48d27fda931958693c42780d070c44f4ff5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d66f48d27fda931958693c42780d070c44f4ff5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
