)]}'
{"id":"openstack%2Frequirements~805959","triplet_id":"openstack%2Frequirements~master~I99743445d5a96515cee0c2ce88e1650aacfa4901","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I99743445d5a96515cee0c2ce88e1650aacfa4901","subject":"update constraint for osc-lib to new release 2.4.2","status":"MERGED","created":"2021-08-25 07:53:24.000000000","updated":"2021-08-27 02:05:19.000000000","submitted":"2021-08-27 02:04:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805959-new-release","meta_rev_id":"3177d2f217b12f557f1ccfab57f5915fcb44147b","_number":805959,"virtual_id_number":805959,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-27 02:04:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-27 00:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-25 07:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-27 00:39:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 07:54:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-25 08:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 09:12:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 16:33:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bf3811a79900545e528a01d8c03cd8db9aa2c299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-25 07:53:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5440ba3f041ff07a6aa8d728f455f64552d78f7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-25 07:54:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"518620b4acf9e71a9981da1f8810f5a0847b8b56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 08:11:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3eac6748e4184a59b6aa519c27d6a16e : FAILURE in 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"cab2eefdec148e80494c1ed92cccfd9edafae52c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:12:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e384f5594f47ee90add6a62f1f8951 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee5401c5c9a4df9bd574c2e54bcbf26 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80d05ce84b684d6ca509e94aee7b2109 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e660f038f3cf442c954dea2a093c50e8 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5ef55dca7d44886b01a79c398b39e60 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3913d6aefa484d3f852f9db5a517736d : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/728cabc919674d6b9bb463a2581bc4d8 : SUCCESS in 5m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e9485fc70c04b67bfd2f7bc2f3acefb : SUCCESS in 11m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1e09c7a66edd4fe897a339ded9041c9f : SUCCESS in 11m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/da62cd65c478440c992cb64a9958172f : SUCCESS in 6m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5000c35a831449a6852cf7ba0e435f3d : SUCCESS in 10m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/824d31bc762745cab13b96aede2c124f : SUCCESS in 6m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c0a5efad2bef48a0b84e191d235762bc : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fe14f2ce9d94376b289e70c4c3aad56 : SUCCESS in 11m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6e77cfd5a4c54e5eb6f1c73b429cf452 : SUCCESS in 9m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/538a095a853746d085788ec61708a6ab : SUCCESS in 16m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3644ce89f4fa49478f008f7ab5ca8124 : SUCCESS in 10m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f390489678542888159e37efd5453e2 : SUCCESS in 6m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4dbe1627c48845689341549dfcd3d147 : SUCCESS in 27m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8490a1e1935b4827954e28ce8806a53b : SUCCESS in 14m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8f81646710c40aebabe20d44934142d : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b43748e7b904918bc63d596fd9731a5 : SUCCESS in 18m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b0448e2ccf2b4dca843b101f669bd56e : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2db84512d9a24c338bf410785c53cb4d : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5f866d74a5334e68ab648e476c7885af : SUCCESS in 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a97879336bb45f785b45b8f23e312ef : SUCCESS in 47m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51d71691fdb248279a2da3587366b3b5 : SUCCESS in 1h 16m 37s","accounts_in_message":[],"_revision_number":1},{"id":"eacc9dbe71d54118d6bd491b6e4ec67ac927b46b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-25 16:33:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c90a6b4fe9e7e44cfb0c8e22252f4dfbf816438d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 16:34:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"353de8ee2c63679a577e0dfa942a6cbe4b5466d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 21:43:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f89fa649194f8795366c2ff9b7adb1 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8d7b94407343539fe2e319af78f245 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/294f070e840c4210a7e4760ebc848310 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cce3ed91ded4410a230ef156cc82207 : SUCCESS in 6m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/658ac801949e4e279757ff445a603774 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4addba390789442583c50eb33c829b80 : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d4e8d3da44c147019a9d060da74dba7d : FAILURE in 6m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5b99e0114cde48cdacfa2e12f6b940c8 : SUCCESS in 11m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0925336fc3f747b4a1db407a4b92c0b2 : SUCCESS in 15m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/30fc617093a049c39da80aae83366cfc : SUCCESS in 7m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e35b2012677b47b1a91b195c87e2bc44 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8cfc06a1deb844ea95480c52adc4f3db : SUCCESS in 13m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9857a59dbced4569aaa3f1dedbef26df : SUCCESS in 13m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a00f234cdc314ab48296a6d59c7bbbba : SUCCESS in 21m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1f46e28c662343dca402550254e9642e : SUCCESS in 12m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b569d28c0c334a359809a959c05722df : SUCCESS in 8m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/14b8dca4c7b7441a88c3abe1ba854cbe : SUCCESS in 30m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2d76428ae92648a8a723c66a2ea1b2bf : SUCCESS in 16m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/01452097732b4c06942fd4557ec46372 : SUCCESS in 8m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21854a261efa4aaaa9ecae5809c89006 : SUCCESS in 23m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5e750f8e82594953acc8e9c2d533eaee : SUCCESS in 8m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/135017417c4b4007a0718aa0cf1ea6d8 : SUCCESS in 9m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9d34d7156a1644319ac26d3d5e386f1e : SUCCESS in 16m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/671df43a221d44969575acacd52e1f34 : SUCCESS in 43m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dacf18f221b445186387f2ad76fde14 : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":1},{"id":"466ed93c8d7232d68b490168f07b68c61c126b2f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-25 21:56:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7b4fe3d96db92bde64c33202a42af162a6fbcd31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 22:12:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c589e50af1784a2da3fb3273bfcb97b9 : FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5ab32f08fbc77d030197ae875087890ea7290ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 23:25:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276283e7a99544ec96508018851fbf57 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b320039206d64984baf39ea6b7c780c6 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66540965bf004c0b8dcd59b0800063e4 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e19cb89b6084b098cbd197c7013541c : SUCCESS in 8m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71c1745ea01e4e9fb7f9e2f6af876c0a : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87c604ea650f41938ec874b7c9a8dca7 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fc9d290f43943a89de179ec646663fd : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/82763d1122b3430ebea71f0eae56ff90 : SUCCESS in 10m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/598a0260c4b5463ead34301ed55d7b03 : FAILURE in 4m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1af54f71d9c54c008d34e66b3ff82726 : SUCCESS in 6m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b06c0dd3b7ff4a9bb3bc8a4ab56c6802 : SUCCESS in 9m 51s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/eff396fdc6284d33abc95d69ca14f74c : SUCCESS in 6m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b34028aeb98f4b51b2370ca79d48d45c : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a40d03e6deb4072a4638aac49fbbb73 : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0aa67ab5affa4da6a2748dc0dca09b4b : SUCCESS in 9m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1d4efc1b7cc342aa8def0d69c7c8d00a : SUCCESS in 17m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a3cb7e1f74654943891e435bb7f9e424 : SUCCESS in 11m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/51afc28210594e2a83ac28de734b2ac7 : SUCCESS in 5m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/637ccb0392ee4c92af1b0a86ef6a506f : SUCCESS in 21m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b44be38ffd8d4002a0876071ddd43b92 : SUCCESS in 14m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2f9fb90aaa04e52a8d986104eb5d81c : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22cfa441217c465aa929ff302fa88d81 : SUCCESS in 21m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4f7b49a8e5f143a2bf009da49328b87c : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ac159d40cd048f480793f6c869f02e4 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1f5a51b3aa2e456891c58349dcb52590 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de855294b5724c7cbb0a123f44d7889a : SUCCESS in 47m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12aa1fbdc1f34721a105f021d9bcc932 : SUCCESS in 1h 27m 26s","accounts_in_message":[],"_revision_number":1},{"id":"275caab5a170a11554a57429428f6c4b21a06ca0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-25 23:27:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc667e2dabeb9267c22014e91839b4d867924225","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 23:43:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4821fed5d3b0474896790a2178eee632 : FAILURE in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2f804fd0c7c801baef6ac17a5cf1cb8e62284046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 00:44:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0ec07dcf679411bb80704a1cb7faa01 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7608d3abca34bcdbfc358614df143c5 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/165ad23b6dd74ab5a4ddd55fea1dcd84 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbec3520ba8b4a5b9c1eb8763e708261 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75a6904e461d4b39b1e93512278428b0 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/282cd4c4a50048e99f9a605c8fe75092 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ca8d60f119a4b13bfa9ebf22508f894 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a33b2efcb3f3464ab70a570736a526b4 : SUCCESS in 10m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d78314e39ca4f0ab27747e78b8a25cf : SUCCESS in 10m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7933ba828e954c2b9f46239b0158488b : SUCCESS in 5m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d8b9cc1d0feb42709a0d8da81e119f01 : SUCCESS in 9m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c41922755fb346c2aa3c78460f1b8f1f : SUCCESS in 6m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a4708c4cb05475b992f5c420c94ed46 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/579b43a3808f40849bf1f032b4da0613 : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ebd7da11d42740389fc958d6bc22a3b0 : SUCCESS in 9m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fab53cc953394ba29da2b9f018049d50 : SUCCESS in 16m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1ef4e19089dc45058f98ab76fd54794a : SUCCESS in 8m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dfad4975baac48b988eb0e3ebe9b3033 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/490ab8ab1a7f4deb994243d02ab4f99f : SUCCESS in 26m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d24226fa95744365897dc9da3862ff09 : SUCCESS in 15m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/66b2996729d545439e089e78ca71452e : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54347cbb2cc54c688457781c9c4b2ee6 : SUCCESS in 16m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/da513b6478bb49b4a3246e1f07bd2ff0 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7039808f4528437e80ec94ad87c6581e : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a5a2ec6948bb4c62b82b524f0d1ce558 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96defabbaa334ebc8d3d7fdb6f2e16cc : SUCCESS in 44m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40cc24f9a6d343bf8b7465636076d9de : SUCCESS in 1h 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8b6557abda34b036ea0f076a7f1d05742e8451fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 00:45:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddb9fcf9b3487051b6d9c60a4d601e1f696b0336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 04:45:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad47113d3610436eb792f470c8e982b0 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a9a3c5ec684eeeb60c3c85aa411c4a : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7463a02fc1e24dba8bbbc2630fca0a50 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4564793ea6f4652842143045b78b8c8 : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/304b4872f991468ea3fa269d368ce761 : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/167b3687f2a049baa8c9c68d486241a5 : SUCCESS in 5m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9df447b8440e4acea3dfff6a74ef1062 : FAILURE in 4m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03b2754c6cfb436e9cf958ef67bf5332 : SUCCESS in 10m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4351242cc7614f3c90ca86edbf00865a : SUCCESS in 15m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5a5cd95666dd4282adbc92e4a6c2c480 : SUCCESS in 5m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4576aced525f4c279f86e105525ed975 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad9df3f207334dbebc0cc7cf45226ebe : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3abe7246a9cb49e093a7eee5dca7a079 : SUCCESS in 10m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6d775a03a58e4e7abc5dd46b662d72e0 : SUCCESS in 14m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d620c8fb76a344d5a55e98079746c92e : SUCCESS in 13m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/18a5e37e346b49b6afab8b344cccde5f : SUCCESS in 7m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4dabc9b7583e42b0919549ec76fc684b : SUCCESS in 22m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dfb47560a2464c39bebd04d68eea08c3 : SUCCESS in 17m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5238e2b1038348c9ba3b0e206b01985b : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5cdbb00ae28c4b8eb27e260f1888b7d1 : SUCCESS in 15m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3d0a286ca8c54a3ba707874dc5936076 : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14a47b9b37a64ebdb2dc677c23ed204f : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/aaaf69c8e5f1459d8bca7c6f97689626 : SUCCESS in 17m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fcf3c8758d34f128aeb292de3fc3f28 : SUCCESS in 36m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87c52700bd3844a7ac6e34dd42609481 : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":1},{"id":"92c4b109d04df10f780b5c27ca1dfc2933543d23","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-26 13:32:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"43b30da8c4584f67a203953ff7ca8c7710eeed57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 13:50:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0d59a4ef5e734938b7ed94d4f1ff4b05 : FAILURE in 12m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4a2322bb34767a652cfa79a0b9bd110a8c088b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 16:55:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd6bebf45724f868c9898de7df07d49 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81e98cd443f74be4acda26d9c53f40f7 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721d76505ef848feb24c57c478cfed6f : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da71de020580422d8f685003f2b8743c : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6540665f8ba040809bfb6c1b15c0fb3e : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1f87440abcd401d801efe7b240a413c : SUCCESS in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9958c768a1d14f34ada57c123bca28b9 : SUCCESS in 4m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a2435d292ec468ea970d4ef4a8dad47 : SUCCESS in 11m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/477d0e72b8df4718ac4a7b481dc2b1a4 : SUCCESS in 9m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7a603065d5e74018827ecbe90799d99c : SUCCESS in 6m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/172bbf1beea544fc8a77f9c95af0be27 : SUCCESS in 9m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f3b28806efa54c048f9548a0d1f6b05f : SUCCESS in 7m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a3c009292ad24f9bbd5139a0dfc671cf : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5411108108494e27812e45ccaef50203 : SUCCESS in 11m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e61cd75983d94cb1b1904b270ff31440 : SUCCESS in 9m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e3dd18c82353417fb11065918867a79f : SUCCESS in 15m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4aaafc7175654a1f835f64175dd5f460 : SUCCESS in 11m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9fc49cc9e9154ab1a2405250afc86299 : SUCCESS in 8m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e410b71ca14f48da8671a4506b429f78 : SUCCESS in 22m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e4bfae8c3e8841c39a02097e4badf70a : SUCCESS in 15m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3406b3ae04954b00b097a100c3e997fe : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ad05d346444486db56fcb6b7e03c51b : SUCCESS in 18m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0a2e7a067b3f4d72adf0be8f8288de55 : SUCCESS in 7m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/156f907d664e4ba1afaf52a605831bd9 : SUCCESS in 8m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/afe18492ef8a4230ad1f80c2df2c0da0 : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65db8c1466db42eba8af5769197ea9ec : SUCCESS in 56m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69258e3f53764ddc83fb298c746a9809 : FAILURE in 1h 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cd33102632c5fa7e04a02968ccbc7eab1a4833e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-26 18:02:40.000000000","message":"Patch Set 1: -Code-Review -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3d86e3491646a4aef4e0bcb9a06237bddbb53f21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 18:20:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3762f733f93c436d9602486ef9f1d308 : FAILURE in 16m 32s","accounts_in_message":[],"_revision_number":1},{"id":"50c791586918cd62c0026d05d3a48e7d7b96c82d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 19:54:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9df9640851442458a537069df7c492f : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91d19861d5149d5815cdd891a87108a : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad185ea6272f45189b9180a83fc42c10 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91dbdd6596934f75a5d672ab350f558b : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/058a21fba5ed4c4e965517c334185708 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5529794d5c8847eaa6bd63387ac423ef : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e98cec895234e58950959e3bae4ce44 : SUCCESS in 5m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/acac41ce86e24ebbb670d7163460a9d5 : SUCCESS in 10m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35c26327aa5d4ab4bbcdb2de322c4fbf : SUCCESS in 10m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/efeca73550a941d1aca8242ea7b353b9 : SUCCESS in 8m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1124c834d4a74781894b677f74ae19c3 : SUCCESS in 12m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ec49ba7115f64124ae62185e0d648072 : SUCCESS in 6m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6c312b4159e9433a9a007ebc98910eee : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9702f2c508cd46c6b7da7c4018e4f118 : SUCCESS in 11m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/479671031404432c89e2cbb73dc095ee : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/09e2762912114156b9e598c7e773efee : SUCCESS in 15m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4899a70f22384137af4f58f4a07215c8 : SUCCESS in 11m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/536a060733c441ab98ab74f341f94048 : SUCCESS in 8m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a19db0337f59480a9cff23ce382ad9d8 : SUCCESS in 26m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b54807169f6945f6a0812948ceadacbc : SUCCESS in 19m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac9191e9c0424d39a092c89f7634aa50 : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6894efe3a0e942dca58b38809ad04531 : SUCCESS in 18m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1e19ee2e5f79478ca80de511c33aeccb : SUCCESS in 9m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9218982446d24bcd900c7f334dd43a5a : SUCCESS in 6m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1f0813a0621641e79d2d42dc92263c65 : SUCCESS in 13m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf42583c49bd4b20b4f346de3543e8be : SUCCESS in 44m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8fba99e28ac45df890b058780399870 : SUCCESS in 1h 19m 30s","accounts_in_message":[],"_revision_number":1},{"id":"507b0d7ca563076db53ec87aedde49791e61bc7d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-27 00:39:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d552249e82fc36196004ee63ce31cf33186633f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 00:40:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3e1497be4867f297618723637c5a1e90f6e9216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 02:04:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d249f9da9c2412b9a4108c3c66cd360 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d423fc53baa4e10a4d739378752b323 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76bed23344aa498abf70fc6718aaf908 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1a2772ddf894e68a487a1ab73add98e : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03e43d49d435494689049c616c050f1a : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b9c4b12f5b3499eb33954875750f36c : SUCCESS in 5m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/512610b66ebf49499797749678ff8937 : SUCCESS in 12m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dfc74fd1570d42c6b48b8177e21491cd : SUCCESS in 12m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2fb2c420a9dc4e3893297579f55e98ea : SUCCESS in 10m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8023fb4f8d694da6bb41cd461a1244fc : SUCCESS in 6m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0bbdaa1210d241cbafd04843a5910d7d : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/33beec17bee2498ea6332858fbdd32e6 : SUCCESS in 9m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2b2715fe6be74bb4b5bd49f3752b00de : SUCCESS in 10m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/361a414ad4364346984087fe40b7291a : SUCCESS in 16m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b0cdd7b39ea54cada95c80142f4e44c7 : SUCCESS in 10m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/508101472f9f41a4a9c70bfafaebce2f : SUCCESS in 7m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9717b572ca444062af7b5a76d520dce3 : SUCCESS in 23m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f6ba658096f94a6bb0d6ca779d52ca0d : SUCCESS in 15m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/811a940a22a040eea54f8d7cc4ffa777 : SUCCESS in 5m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6798bb6f8f684369b466b3cde40558e2 : SUCCESS in 18m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5e0f9caa227b4118a65c8397e66f3a9d : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2511ecacfa934776b4081cca9dffa81a : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e73f37918de1434a8ed193140f7e2fae : SUCCESS in 15m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e4b871b9da949aba6423f493008e0c3 : SUCCESS in 43m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5ebe3246b2449148cefe89bd1a26ca7 : SUCCESS in 1h 18m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8d048aecf45b8618fc13110ece9e69edec804c0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 02:04:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3177d2f217b12f557f1ccfab57f5915fcb44147b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 02:05:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6f71b39699494cbed72a1a7dd07d27 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00d7d9ad65f2a211af1ebcaca88e9fa37f58237c","revisions":{"00d7d9ad65f2a211af1ebcaca88e9fa37f58237c":{"kind":"REWORK","_number":1,"created":"2021-08-25 07:53:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/805959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/805959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/805959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/805959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/805959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/805959/1"}}},"commit":{"parents":[{"commit":"26d7e48ee0e585eab06043894d8c202123ace701","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26d7e48ee0e585eab06043894d8c202123ace701"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-25 07:53:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-25 07:53:22.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.4.2","message":"update constraint for osc-lib to new release 2.4.2\n\nmeta: version: 2.4.2\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I5b43120f291a14e0c37f8fcb9199947399f6e5f5\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I99743445d5a96515cee0c2ce88e1650aacfa4901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00d7d9ad65f2a211af1ebcaca88e9fa37f58237c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00d7d9ad65f2a211af1ebcaca88e9fa37f58237c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
