)]}'
{"id":"openstack%2Frequirements~805962","triplet_id":"openstack%2Frequirements~master~I2e31340e993f4106f910882a22ab1c90174a5d84","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2e31340e993f4106f910882a22ab1c90174a5d84","subject":"update constraint for mistral-lib to new release 2.5.0","status":"MERGED","created":"2021-08-25 08:05:02.000000000","updated":"2021-08-25 21:44:32.000000000","submitted":"2021-08-25 21:43:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805962-new-release","meta_rev_id":"1f9fb62082084da1d6c50ac34e0baeb9efb2ad37","_number":805962,"virtual_id_number":805962,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 21:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-25 16:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-25 08:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-25 16:33:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 08:08:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-25 08:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 09:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 16:33:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3daffcca1a9544c0e430b1bab2a84dc4d4bd46a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-25 08:05:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b222b1545ef8d8ee7e92384c04d2e17b46e3d6fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-25 08:08:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"959e1a20b69033e4fb5f6354278484e4729d3678","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 08:20:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e542d6e885f642eca6ba144935da356a : FAILURE in 11m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6956ac18606a4fbfeb03c98282a0ad095069911c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:34:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c7ab78bcff40d8b213cb0ed3c62a9a : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fafda2d2a734198a6a709aec14a22f9 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27653f72065b4cd3a6291ae012f4a05c : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ad46a69abfb4a8493bf5dd727e6f7d1 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69ec9f5c6a864782b65ece7fa56f28d4 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2fe106ebb58b426d9229cd1b7cf2ba7a : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3b14e022ea54ca9a88ca6b138ef28f7 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2942b4620d514886911a21754fd78024 : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2bbd961b3bb74b9d8da4fe098affcc07 : SUCCESS in 10m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e37658a868bf4735ab6b5704b743ccf8 : SUCCESS in 6m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/58c4cebaeb00411197fcc3e05d47f67d : SUCCESS in 11m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4516ab9e182d4917ba419de03ecb6938 : SUCCESS in 7m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d73f12a5562b45c29d1bdbf30e507519 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/585cef2462184b1bb9933575e261f1ff : SUCCESS in 10m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f28a52f6ba2040afb04148332feed133 : SUCCESS in 8m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/923ecb715f7c400aa7b117b18a29cf1e : SUCCESS in 14m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cc5517817fef48dc866574f5f93f3088 : SUCCESS in 13m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/aa68ea3a1ac2471b973c1977d5160d44 : SUCCESS in 5m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/49853c4856d24e44a5afc8002034b271 : SUCCESS in 28m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/03b09e04abc84461abba0aef3233dbd7 : SUCCESS in 15m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de50118a07d1430a92de63ccd9688feb : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3dc17dde60714b3c9e7812d81dd793a3 : SUCCESS in 17m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fc5634c0a3a8456fa532b2d92b672d7c : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dae6a6bc183e4dfe91a12c89ba1dd3ac : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/890d91843669466881808eec290e8a81 : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac01850694074ce08442643e81ccca26 : SUCCESS in 44m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2eb85bb21ef49f092eddc723eb2d2e3 : SUCCESS in 1h 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3ab3a79cba57e4c212c755ad1d6752ecafb65f25","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-25 16:33:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91efb8015511c248f88dbeb5cae7922d6ab3f666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 16:34:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac33faf20d29f5d6840daeb88f7e070fd43707be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 21:43:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b37422d0d0f47da9ca62050b1b79a42 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2190ac5e641241089c2d390595461224 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e2aac89efd04caab949f40fe3a7538c : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18a62184042c48df8d5f2a5dd281d987 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d60748a9e31448d086ec41a6e333218a : SUCCESS in 7m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5d8af1633b84066b438cd63fa115c89 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7853b0ef31674823b2e555b8131806c6 : SUCCESS in 11m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/275bef8545c44d8aaee12d9310ea7071 : SUCCESS in 10m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0ef3aaeaa93047a6a15a78e43b14e4a4 : SUCCESS in 9m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9c648fb1e6824734865b5ae819af00f7 : SUCCESS in 6m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/070cf8115d7145c7b41e48482a62dad0 : SUCCESS in 12m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb664e4697f34a77943118174785fe92 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c545aaf5b13b499aa276053d0cbb9953 : SUCCESS in 9m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e897fff757ac4abcab59d2a1961156f7 : SUCCESS in 14m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/83b9224dfe194883a12ba30972a202c6 : SUCCESS in 10m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/15835216cf704996856592f8ad1d8cbc : SUCCESS in 6m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/aca99c2069b847eea5e6b7e2da486027 : SUCCESS in 23m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1574ace811c74844998bd9ad3ea84e36 : SUCCESS in 15m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23c9521574ae4886a533c2e9d2212a9f : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f873042b81cb4ffcb9e52b08b568b450 : SUCCESS in 17m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e7a840487a94470bb0efcbfe9272bcd1 : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/903de95ce2974c31905265b09cab2202 : SUCCESS in 8m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c61f115dc3624c548fd4be4228b2eebc : SUCCESS in 15m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5534061a6bef422c9be9f6ad5b731cbc : SUCCESS in 46m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5ee6c535aa043888d8de4aa958e7909 : SUCCESS in 1h 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"83ef19d9325fcb2176d39f2f5170e3498af4862a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 21:43:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f9fb62082084da1d6c50ac34e0baeb9efb2ad37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 21:44:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9074ce83c1ee4d2bace0cbf1aff9f4f4 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"348d77c6753ab636ba22f646fac1eff3e486fe47","revisions":{"348d77c6753ab636ba22f646fac1eff3e486fe47":{"kind":"REWORK","_number":1,"created":"2021-08-25 08:05:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/805962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/805962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/805962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/805962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/805962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/805962/1"}}},"commit":{"parents":[{"commit":"26d7e48ee0e585eab06043894d8c202123ace701","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26d7e48ee0e585eab06043894d8c202123ace701"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-25 08:05:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-25 08:05:01.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 2.5.0","message":"update constraint for mistral-lib to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie8592c03aaf4b82f9159333b222cd78c23a430c1\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2e31340e993f4106f910882a22ab1c90174a5d84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/348d77c6753ab636ba22f646fac1eff3e486fe47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/348d77c6753ab636ba22f646fac1eff3e486fe47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
