)]}'
{"id":"openstack%2Frequirements~806166","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I784b24e2fad176c61d7b067d97f71ad43a2b631a","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I784b24e2fad176c61d7b067d97f71ad43a2b631a","subject":"update constraint for ironic-lib to new release 4.6.2","status":"MERGED","created":"2021-08-26 12:27:52.000000000","updated":"2021-08-27 01:01:48.000000000","submitted":"2021-08-27 01:01:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806166-new-release","meta_rev_id":"27bdb5eed7508de8e514ce26a280ca3a8eaa3fbb","_number":806166,"virtual_id_number":806166,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-27 01:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-26 15:19:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-26 14:15:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-26 15:19:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-26 12:50:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-26 14:15:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-26 14:52:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 15:19:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"85abd9546faf87e26714ca4c156bbdd9fcd26b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-26 12:27:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b5d7e2c37a3c03603c2ea41994de0403c9aa044","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:50:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ef9de3994daf4f4aa8c072b323cfbd3c : FAILURE in 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"32a307df467da275a31b349b4e41df6f888b449f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-26 14:15:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d5e0cb9acd0c222991049c2b845419fc70842e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 14:52:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68725a490a274ee990f1e89cb2ca5b9e : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/964fb9dd18d448d3b295e44c478477cb : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dadca5f1c1944c6d92a8825bb9460637 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ad05c37a1b74c6988a89f410543ffec : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e57e3ee542a48b2aab095b17cd16d49 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2838a6d647e4debaf3af465a1dc3ebd : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8838ba7ec63442abb87caada3ebe4b53 : SUCCESS in 10m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28cc993520824361b23cf5d965544fbe : SUCCESS in 11m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/49c50f6d0b6148f6b3fd8ae1b3b0c5f9 : SUCCESS in 11m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/49cafc504c424850841d0d5cabc866e7 : SUCCESS in 7m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c1f4dbc6bdb54c3188f672eef8e8c8a6 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f3aedca8fa44fb590bcd27e397dabd7 : SUCCESS in 11m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f560774820ea4604b57cb6f737ac26b7 : SUCCESS in 10m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/04d90c84930c4da3a2a2e561b3ac081c : SUCCESS in 17m 40s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/dd7e97ebe6794c6e84bf35f877ca9918 : SUCCESS in 8m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8e9c20a7d744b798cca8249c47ef589 : SUCCESS in 25m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/83788fd8cd5a41839789e37e9f61918e : SUCCESS in 14m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9fba91ba202b4f36a560435cda092686 : SUCCESS in 5m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b863e816ed540c68ff27a53091c2963 : SUCCESS in 19m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ce16d7312d534648950b7dbb8ce24477 : SUCCESS in 7m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42ea30434a1d430d94086dd394fe53ed : SUCCESS in 8m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2a2a38ff6825486aaa147709f2b1717e : SUCCESS in 14m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe674eb0e95a4bcfbcb01aa2262c3619 : SUCCESS in 52m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09b865cc21e348498dc0194ab9a64d25 : SUCCESS in 1h 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"668585ac132b4dc5e9047c1215524d4b3c0260f4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-26 15:19:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d796121c29f939ac86cd79417e2b4bd756383a77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 15:19:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8b69ce8b09c2811f2ac8cb1393d079a9b91d66b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 20:53:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42b9aaefe654b7dacfac71582f5619d : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01ee1a366c714d0f89881f5723ed8701 : SUCCESS in 8m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70102819b27742df8700df70363e2035 : SUCCESS in 6m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07ba41b32e73490880bf53b586e1e1ee : SUCCESS in 6m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4985e4ef2cb84b6a9cc89f9633cec95c : SUCCESS in 5m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b3a960704b864bb38f96c32ea72e3ec6 : SUCCESS in 10m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f997f935e00e4c3a8602cefce2637a16 : FAILURE in 6m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/31b1d9a12c984067b72c33e9a54d397b : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/724e3016940844edbd23df99a4fea88b : SUCCESS in 7m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a9f3b164adda4919b9c44743d63fba0b : SUCCESS in 13m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87ed0d91f2eb45d79b5633c39ceb6ee9 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9797d476976640c481fe47d4d5676774 : SUCCESS in 8m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fa8c1df6b57540f2a1b740b423b9e527 : SUCCESS in 14m 32s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/dbeac15b4d0e4f0eb58bce2170bc588c : SUCCESS in 5m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/190507ca95de472b811f4ae55d4f6c6e : SUCCESS in 24m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/918737236c1d4033bf72fd8747496264 : SUCCESS in 13m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/be7868578a61484aae89daf692d5a9c7 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9dc2d928ea74e86903b6b64f0460d63 : SUCCESS in 20m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/da47208414b742c0ac52106f58156ef0 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e5a4eea044943aea589658fbce6c546 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/10eef2708e084ba4870331552ca8863a : SUCCESS in 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e04e9153bd0f4a678c0c4796373b4644 : SUCCESS in 50m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6620d979e31a494e91c6883154db38cc : SUCCESS in 1h 25m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e0457187f49b489a87b970ff954f3af3e3e53ea2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-26 21:16:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03334be34ae8415af48d61a964b75c351326b6fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 21:32:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c526b68386f94a80a52fd9e7e63ad909 : FAILURE in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"11f711fdf02c1ab84b056644f6812a49d8d8d180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 22:57:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a365933253f47bd998d106014651e6a : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92f774583a1844909b87122a63714d20 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/066bbd2b407c409b9d0b8cfedacea518 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3589532bbe6411db046b8d947866437 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab8ca9df6f544dc1bc8730c84426877b : SUCCESS in 5m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9da38f410a134929af8d2d90291cf369 : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb2471c939da44a2b838843e5db03302 : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/415093ce037544ae84e725dc9914b383 : SUCCESS in 11m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e43be4a0e0094c0691aec676da910e7f : SUCCESS in 11m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7e97457b87e04d9db0a3e1c04db4215f : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/658ccd9bf5384df8991fa4eca7fceee5 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07c9692367e94c298264e6b97fb2149c : SUCCESS in 10m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c383055d007144019be602d53d175f8a : SUCCESS in 10m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15caee992aad434c9e3d5b76865c2ca6 : SUCCESS in 15m 54s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/697f52f5589945458f0b889ce1c1bc63 : SUCCESS in 5m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/52839c63f33d414e997c48b488bff8fe : SUCCESS in 23m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ede7ded58d50437dab098c221474109c : SUCCESS in 12m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ab136609cb34f1489e0a8f5d7eebbf4 : SUCCESS in 5m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4457b55691a04584aeef7b2ce5c5f915 : SUCCESS in 19m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/67970b7e55f3464b9532d5e510f4a8ca : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7eed4142bc0c46c0bffd966f4e93a5cc : SUCCESS in 8m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/257d2949d68349e79eca09f376011ac9 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3db68dc3b48447a8a33d010957d3119 : SUCCESS in 51m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71a08f9797a148a8aa905146b3021dda : SUCCESS in 1h 39m 15s","accounts_in_message":[],"_revision_number":1},{"id":"23542e6301dee68e329dc1e5aa2c375f8afa5f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 22:58:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd064b487507fe9e955a0e5f7e0094438e7643bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 01:01:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/196a9d4daaeb4ac2b14b2d506d9abf53 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe1e502e15c24569b29c6bea887dd933 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bf73bb3c67a43f7a5087dce46664f21 : SUCCESS in 6m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87b6e7d71bfb4bb88458ac264dd062cb : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd35740c71af4927bb28ebbc930b4880 : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f630e7daff424aef97ec4d1e7588871b : SUCCESS in 10m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c723bce8ab04676b26d3d22e1fcfb14 : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/796ff248ddd146f98a6b330b94b21457 : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6f7377e8ed1d4bf9ae2c9005e8d38187 : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/93a374a2d564434088b97dda41598fab : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e921b349ea834e25a804e6697a260c6a : SUCCESS in 9m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cee0170eeeb1403e8b08f343eb266ef2 : SUCCESS in 9m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/28366c05aabf4ab5978cf5532155ef5e : SUCCESS in 14m 40s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b3cd446f51704bfbb891e836e90dc08f : SUCCESS in 5m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f75de7a8b7f04496b098b8f047530efb : SUCCESS in 24m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c140b69afc8b4895b4511dbc1dfdeebe : SUCCESS in 10m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46582663c2f94afa98d0f48d84be8dc3 : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6b6ef4a6d234621b8c85eec6b7908ad : SUCCESS in 17m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2b7be8b96ac24cc3aeaf5dce48749f8a : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d4b967ec7c644ad86d303f7cea444bd : SUCCESS in 7m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e0c0ca6bc2cf4135a46e560bbc4aa29e : SUCCESS in 14m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8d8a9917cfc4f12949216cb4b664a51 : SUCCESS in 47m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe24240e757846e9be2b0b9a9acdf4fc : SUCCESS in 1h 51m 37s","accounts_in_message":[],"_revision_number":1},{"id":"27bdb5eed7508de8e514ce26a280ca3a8eaa3fbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 01:01:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56024cd3d6eca0880873098f63e52662dd49312f","revisions":{"56024cd3d6eca0880873098f63e52662dd49312f":{"kind":"REWORK","_number":1,"created":"2021-08-26 12:27:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/806166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/806166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/806166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/806166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/806166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/806166/1"}}},"commit":{"parents":[{"commit":"c3018b1f729be36cfb96d5cd71c9505bf32100e9","subject":"update constraint for oslo.db to new release 8.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c3018b1f729be36cfb96d5cd71c9505bf32100e9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-26 12:27:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-26 12:27:50.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 4.6.2","message":"update constraint for ironic-lib to new release 4.6.2\n\nmeta: version: 4.6.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Change-Id: Ibe1a1bbdcb003470e43309093686ac14037d9d0b\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I784b24e2fad176c61d7b067d97f71ad43a2b631a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56024cd3d6eca0880873098f63e52662dd49312f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56024cd3d6eca0880873098f63e52662dd49312f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
