)]}'
{"id":"openstack%2Frequirements~806570","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I9f53d80bbe20f90d34fe93966b1391bfd4a917c0","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I9f53d80bbe20f90d34fe93966b1391bfd4a917c0","subject":"update constraint for ovsdbapp to new release 1.2.3","status":"MERGED","created":"2021-08-30 13:07:38.000000000","updated":"2021-08-30 17:20:00.000000000","submitted":"2021-08-30 17:20:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806570-new-release","meta_rev_id":"8c9a6624ab0dea91174c476e801cb01ca0780b30","_number":806570,"virtual_id_number":806570,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-30 17:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-08-30 13:29:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-08-30 15:34:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-30 14:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-08-30 15:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 13:29:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-08-30 14:37:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-30 15:08:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-30 15:34:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"eb45e0ce69f537e32e29782746bc146f50e19aad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-30 13:07:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d354f57c28565d91d7188d4aeeeefc9029bcf96","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-08-30 13:29:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2cd6670aa5dcb3c4f1572ad2bbb83c61c5908c8c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 14:37:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8040846bcc62065dc0af34cf7451544a46ac941a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 15:08:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f652f07b0e24eb089f23459235db336 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f529e6688956498aaa8b8b83a5cc8a39 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3782641cb6744182883460fcc73e83e9 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/276ee4bcdf4e418b811f68585988f023 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37d1e3f104994f5fbdc52f81c9336c0f : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bdee8ff8e642445aad1f4fd58bc1dabf : SUCCESS in 12m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b80e6dcc811b4a8a80037821b36db2e8 : SUCCESS in 11m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9490076303e0410f981649e964d9d829 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/757b9c10400540fe9d1cfed0afe9b253 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f07addbde98464dac16b28bfa269099 : SUCCESS in 10m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bf34f52e08644303b01f1c1ae674c91d : SUCCESS in 14m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/112f6162788b437180bf5b6bfe07d307 : SUCCESS in 28m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e45e1064e4a4f38bf0ea164665ee7fb : SUCCESS in 15m 23s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c2f9a9d4212a4a60964e97b90eae3703 : SUCCESS in 15m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76097b31180444caa683b1f5455e51d8 : SUCCESS in 22m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4d475c1143e94fbc8a0c18fa3abee519 : SUCCESS in 6m 27s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/040797b836a9446a8cb3af7674a22ebd : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddc62b8f61d24e36b1aa6fde2fc0b83a : SUCCESS in 6m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ebb8d58f0079450dababc899b0750c0e : SUCCESS in 12m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/48166fb1ca7a478f8d75189c7947c9fd : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de13307aecf64fc0a3494155664fc8bc : SUCCESS in 59m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22fbce8034f74c69847018f79273b6fa : SUCCESS in 1h 13m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/579896c2323a413b85078d8efe171572 : SUCCESS in 1h 54m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/cac5e8295a3247348605001b9ad97b10 : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c5f04c220512483cb1b0f022cd35b72bcfc8df19","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-30 15:34:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36c2aebf9ca1e36260d1bd6eb57bab3ad31d08ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 15:36:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee1adef0b0d0d6fa5b021341ddd9f4b24a43919e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:20:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93df09b6adb14fc1b462887a54ce01b3 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85bd8d681d5a4b6a937d342adc9f669f : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa24ade178ae4abe938f72c82e60cdf6 : SUCCESS in 6m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/395e7b3b9e6b45b79245425041a56e14 : SUCCESS in 6m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92eb111aaaf24d6a9414639319c77803 : SUCCESS in 5m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8df6e2cc9f5d4799a80c153ee7af5e25 : SUCCESS in 12m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f8dc44b5247843f3a997935703d0703e : SUCCESS in 13m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/777465f2a9bc4a2682201e88a07f5598 : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4c8e12d81eee4c9596c37f8a24d63ebf : SUCCESS in 13m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2562311007e4500aab9e9cd63ebc7a5 : SUCCESS in 9m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4c403b736b114dd2809008e17fc391be : SUCCESS in 16m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0714ace4c35447a69db18a636d7a40fd : SUCCESS in 25m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37c30d669a1f4234a5e1ec7a3971f15a : SUCCESS in 15m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/95668279a4914642925dce81f2f3facf : SUCCESS in 19m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/080f68e2c260413993c0f6e58aaadf8c : SUCCESS in 20m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/89f278ea1618496f8a876dca524b87e8 : SUCCESS in 8m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/36b5b553ae6049b08d8af1f8af0c3a64 : SUCCESS in 9m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/14b2463586a6403b8070440eb08f5dea : SUCCESS in 7m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2417c42a2cba4174a1cdf4d94623be4c : SUCCESS in 12m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d7e48082b70d4222bbaf3f70c3fc0891 : SUCCESS in 16m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23435982139e4d07ba57127cf1076d7b : SUCCESS in 52m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/79ba7dcf260e4df9beeb7d455063fc75 : SUCCESS in 1h 12m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/290df4e6346a4cafa9ab7054c3b6a9c9 : SUCCESS in 1h 31m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/95e97f9326a24614b1f3fdc9b3024f72 : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8c9a6624ab0dea91174c476e801cb01ca0780b30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 17:20:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"509302695652c4a0c689d88daeb9cfb1d28e9e91","revisions":{"509302695652c4a0c689d88daeb9cfb1d28e9e91":{"kind":"REWORK","_number":1,"created":"2021-08-30 13:07:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/806570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/806570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/806570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/806570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/806570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/806570/1"}}},"commit":{"parents":[{"commit":"00672c57acd3a2471cf5eebadbcf731822a1e70c","subject":"update constraint for oslo.config to new release 8.0.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00672c57acd3a2471cf5eebadbcf731822a1e70c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-30 13:07:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-30 13:07:37.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.2.3","message":"update constraint for ovsdbapp to new release 1.2.3\n\nmeta: version: 1.2.3\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I0d3ea4cf381cb44bfc7be2503e7e5b3231c1ecc4\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I9f53d80bbe20f90d34fe93966b1391bfd4a917c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/509302695652c4a0c689d88daeb9cfb1d28e9e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/509302695652c4a0c689d88daeb9cfb1d28e9e91"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
