)]}'
{"id":"openstack%2Frequirements~806601","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I223d5e80df1b2c820d59349109c19831e0fc2ec2","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I223d5e80df1b2c820d59349109c19831e0fc2ec2","subject":"update constraint for os-brick to new release 3.0.7","status":"MERGED","created":"2021-08-30 14:24:51.000000000","updated":"2021-08-30 18:12:35.000000000","submitted":"2021-08-30 18:12:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806601-new-release","meta_rev_id":"5a63c75aa411673b024cec401f7da87968fb8ac8","_number":806601,"virtual_id_number":806601,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-30 18:12:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-08-30 16:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-08-30 14:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-08-30 16:06:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 14:38:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-08-30 15:52:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-30 16:06:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ab7608aff6d4db66eef2406f30b6d83b1fe653a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-08-30 14:24:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4c52b0388b87e25da33df5bac2c9f28ff312d71","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-08-30 14:38:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c08ae55561f0b4388fdaeefe4ae150ccdd658c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 15:52:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5fb2d556dbe49b9a32c7c7f7aa75fbf : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/114ab03e0ed64f518689a29a3fcdf6af : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52be31e9ea414d50a32058938eec6bea : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/935235ede74040c3a19103e442eabf84 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/64f57f6e6471473c866c9cd9e03249e4 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f70a7e32b7084f36a18a8733e40b6940 : SUCCESS in 11m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/88aa7bb1549341449724dedafb3c0890 : SUCCESS in 9m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a5e16387d9d340258da78238c0fc28e4 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fdd1432953a04246b6c84248503189ac : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18b02b3a1e9a4254bef128714f142436 : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/18e02ef0ea1b45b696dc4248d15f1002 : SUCCESS in 13m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a58155ded32444138a69723a4d0d1773 : SUCCESS in 26m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/48ea97f900e54093b820eb44b4e192f8 : SUCCESS in 14m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2d9d04a19ae24874894ca9e336687adf : SUCCESS in 15m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c56767af3e34d4b9256cbe62d660dc2 : SUCCESS in 17m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eb953fc772ee421b94f76fad64a9bb27 : SUCCESS in 5m 50s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6c0577b946254cc5aa9050ebc569b0ba : SUCCESS in 7m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39621c0a5b8d484583946d456f81cf1e : SUCCESS in 8m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f9915927f63f44169ce2dcf9cab8c0af : SUCCESS in 11m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ebb7f25d61664179b2eb6f13c38153be : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14e35aa5cb8046bab990d345e9fddea3 : SUCCESS in 52m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f0f3db5934440c2bf062c6ff3079d9d : SUCCESS in 1h 10m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d5fc6423bc844f3a9fb6a64987a82cf : SUCCESS in 1h 25m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/a2c8da5a37a64a6d95bf502633befa90 : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e88023230e77ca529130cf7c88bbb00c73a0ce78","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-30 16:06:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f96677856efa4edd23a7b80ed08d39e094c0d63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 16:08:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ea45f15ffa8b838ff89dce8ccef5313d30f8168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 18:12:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e893bda363cc4a51ba78a9b9006c2cf3 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/542dc74c7bec46bd8eddb65eeb7a568b : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24ee6f638f934cd894d920ff4ab55b4d : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b502fd441a34faf97f23fead5414a25 : SUCCESS in 5m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1db6b0da149b445396dd514596f2d457 : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2b1f7a3cbe04f6ba4d70cc84cc64a7f : SUCCESS in 13m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/164d1ca99e224742bc90d8cfa6d87c54 : SUCCESS in 11m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bd7b628a1137445d9a66718a48919348 : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6c9c846b3f8c4a0a822ae097cfc46e24 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7048416a1b764fa19e5690191fe23f89 : SUCCESS in 12m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/be9e3631452a4eb0899e7cc3dfc5ef5a : SUCCESS in 14m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d7425184631d4f0f8b01f9584e0f8191 : SUCCESS in 25m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4af4acdeaa644b52907b6b8d5d9262c9 : SUCCESS in 18m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/2400c7774b0741cb877f996f04242c05 : SUCCESS in 14m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19796f5aa156400999c79250fafcce68 : SUCCESS in 17m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/742dcaf72c55432e93da26fd46e2cb14 : SUCCESS in 7m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/308f4d014e82447aadf43fcbb53cee2f : SUCCESS in 8m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2821d1ee9884450a7107a66f0972305 : SUCCESS in 8m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bbcaec73316649afa886d8b19bb551d4 : SUCCESS in 11m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d63185f1579b4b6cb559d74d518943dd : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21a8210125ea414aa3f42f08569723df : SUCCESS in 52m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a543a7f27f024892a8fc7b6f55bb12c6 : SUCCESS in 1h 09m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/766026b687e3498588569c9ca501e555 : SUCCESS in 1h 58m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/b838a0b7f8fe47cab6c1d5350bec8f12 : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5a63c75aa411673b024cec401f7da87968fb8ac8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 18:12:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"460ac3244ff19c78c54a297c25e6a15b933da9c1","revisions":{"460ac3244ff19c78c54a297c25e6a15b933da9c1":{"kind":"REWORK","_number":1,"created":"2021-08-30 14:24:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/806601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/806601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/806601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/806601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/806601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/806601/1"}}},"commit":{"parents":[{"commit":"00672c57acd3a2471cf5eebadbcf731822a1e70c","subject":"update constraint for oslo.config to new release 8.0.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00672c57acd3a2471cf5eebadbcf731822a1e70c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-30 14:24:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-08-30 14:24:50.000000000","tz":0},"subject":"update constraint for os-brick to new release 3.0.7","message":"update constraint for os-brick to new release 3.0.7\n\nmeta: version: 3.0.7\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7547ff40737f90aa3a43698e35644e6dfb5a3931\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I223d5e80df1b2c820d59349109c19831e0fc2ec2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/460ac3244ff19c78c54a297c25e6a15b933da9c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/460ac3244ff19c78c54a297c25e6a15b933da9c1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
