)]}'
{"id":"openstack%2Frequirements~806872","triplet_id":"openstack%2Frequirements~master~Iaf000ea45987d29f5d12a3118d8192a9f6511001","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iaf000ea45987d29f5d12a3118d8192a9f6511001","subject":"update constraint for python-zunclient to new release 4.3.0","status":"MERGED","created":"2021-09-01 08:41:57.000000000","updated":"2021-09-01 17:37:01.000000000","submitted":"2021-09-01 17:35:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806872-new-release","meta_rev_id":"cd87612d4f243ab2553ec1546ed0d6c666b0a7d7","_number":806872,"virtual_id_number":806872,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 17:35:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-01 15:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-01 15:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-01 15:32:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-01 09:03:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-01 10:52:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-01 12:53:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-09-01 15:32:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-09-01 15:35:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"324a67386e2b54a0672b24792adb5070034f1397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-01 08:41:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70f3bf951aed789e7853e34f9fef4b88b134403a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 09:03:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6d8315c9ec0b4192971c5da98d1f3c61 : FAILURE in 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"00bc355a0dd7e991c70526308f21b7dca5dbde60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 10:52:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6bdf5db75d432fa1bbe6660d6b5f96 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ab27ff90446405a893bb6c90870a32b : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b751e8f89034add80bf1e907192d172 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3914eac3aad4ccd85ebdfead0b67606 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/569d7ed013894e8884e5b5daa951d003 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9593f5308ebd48a3bf918975c208085e : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e8d8e08651f49d892dbdf391becbd39 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/81acb6c6fb3a4ea3879c9146bb79e957 : SUCCESS in 15m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3987a100dc2b40839dfd8f329fad5cca : SUCCESS in 12m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/df5d9bda779d4dc6903489b3e4743aef : SUCCESS in 6m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c4f5f232fa6a4f9b8b2c57ba97db6398 : FAILURE in 13m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fb28042355314a03ae0e09d3b2356a84 : SUCCESS in 7m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8fd944aa804544f19043369db3b59bf9 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8dac28d9652b4120b01ab6b7015ffb56 : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bfab8eb461ca43abb39e81cc8fbb078a : SUCCESS in 10m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/757afacbbd354634821d826736b6454a : SUCCESS in 14m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2ff5114b94664faabca1016361d49650 : SUCCESS in 9m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5b068df9cf784f50903e9c4f9daa7ec8 : SUCCESS in 7m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/57e776d642e84f90917e52955985add0 : SUCCESS in 22m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7f49f6485b534cef9f995bf206cbf2ed : SUCCESS in 15m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed7fcd0701524fa59093d3ae62d126ea : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c1a27eb2cd2448a39392a7f6a6855d6b : SUCCESS in 21m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8e15ca08a8704ce0b0a5655a3b704b72 : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7d5f370e6fb4bba8fc43e0956aa5d84 : SUCCESS in 11m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dcb2956907e54d82af96c89690324996 : SUCCESS in 13m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b0b0dea60da48f7bc5eb82f8dd1f656 : SUCCESS in 43m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fe1f2f7d90c49d09b342cc609387de2 : SUCCESS in 1h 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e4e8cbede3709a19f0b0aed4c27bed0b31ec1239","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-01 12:53:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"14d24b6dd2fff9f7822936cc4372dfea4627d87d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 13:09:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/55d4213848be42fcb1d83846bcbce55b : FAILURE in 11m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e2e4baea7932d9d219d5bd32ee06a582902204cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 14:10:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb7e4cd8831b4f4ebe4e0efb5055460b : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0008e5fa9af142b7ae1040d400e33706 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba369d2f64343cb964a4b437446b460 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9379e9ff9b3549d7a78e6e4565a651b7 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fc777ad4c194e3c885f2f9ba999efd7 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f19ff30ab724940bcd04c94b44b2d05 : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e3cd76987754fe391a7bf2e68b08255 : SUCCESS in 6m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/576ff3e5b79145f19dd97f06c14d6606 : SUCCESS in 13m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8ba9ccb806f4b38a5cc219113e6de67 : SUCCESS in 11m 43s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/402aca85ed3042bf9d0d66b36e1fd13a : SUCCESS in 7m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a718ba5e523e41bea84629d9b4465f8b : SUCCESS in 12m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5cb96783ed544f978cf79a04e6cc6f1f : SUCCESS in 8m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/05efe14775484649bf7b02deba9c07a5 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3228d4b464f148e3aefdb464fd3eb304 : SUCCESS in 11m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d9159eab159b4abdbf61bed97a761ee2 : SUCCESS in 9m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3a39d8d169794ee1acc770c93d8d2a42 : SUCCESS in 21m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a5aaee15b90b406ba2ea8ec8d86ce101 : SUCCESS in 13m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/986d58f17d09406aa88cf2f5ebfa3fd2 : SUCCESS in 6m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5cb917933af143c1a71d448da96c9f4c : SUCCESS in 28m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fdcd774a184c412ab9a5c10620d2eb04 : SUCCESS in 19m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8bc31b5b1d0a4d0ab5d00d854ae1d332 : SUCCESS in 5m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e952765edbf445cb1c52e67d46a574d : SUCCESS in 18m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f5e87d755e374ef8a6e925db0eee3c93 : SUCCESS in 8m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d2a94a36b424a70bc8d2edc40d6c2e2 : SUCCESS in 8m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6768c2de852d4cf083d79c138bc4f593 : SUCCESS in 12m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/969a0cd7344b4462ae5ad55edd274c31 : SUCCESS in 55m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0a2cf475e954c2891a4935d89dfb550 : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"537733fbc89f535fdad37bd83da52cddc05aefc9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-01 15:32:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"334d9c94c8a33e73c05b842f119d188124cf3611","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 15:32:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f49a49c013ba358c033c835ce68720f3561359e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 15:35:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a51bc76b3aac5a7dd8e8f75ffe209e5070acdda7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 17:35:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f3df8ee18a448698a2f514ea8a6c2e : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd16ff1a657343c7a654ae09e21a463d : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe037f30e3e44ce7885f497d844f8e77 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfdc4dcdee5b46828d82ae6551e1fb23 : SUCCESS in 8m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b1b131d211b646be83f8c213ff321386 : SUCCESS in 7m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/02394584a27346569fa7c3545aab7275 : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a4f94e13ab214e4994da8d389d4bd5f4 : SUCCESS in 11m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8f2d2eea9004f0cabb028c034dbe205 : SUCCESS in 11m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c542dcc47b0948d180c2773ee55680a1 : SUCCESS in 10m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cbdb3fad190746d1b4f2a0823655d8ed : SUCCESS in 7m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5b669310f48a463eaf5e467a1dbf82f0 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45a95383f80842518df8dab0edd70a82 : SUCCESS in 9m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca0198e8cd474a049ed5a1ec7353569c : SUCCESS in 11m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/48549286ce7a437ab1cbdb19129c122e : SUCCESS in 15m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c958f6c4c3a341fd94c40fd6511ed125 : SUCCESS in 12m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7f3390e6cbf84d1893b76ce054d25ed5 : SUCCESS in 7m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/78dcae714a744421a6f46810b6a77e6f : SUCCESS in 25m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5fc5d8c9556f44818b90ffbf9131681a : SUCCESS in 15m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afd79ca7fc084f128abb83523ecedd2c : SUCCESS in 5m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e40dccd9c89a4ee886b4f02ee0958748 : SUCCESS in 21m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/729b7304cbe240219f20cf8b626f5383 : SUCCESS in 8m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c07f5dc2093741b5a28a63f4fc74d0b5 : SUCCESS in 8m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4809754294284dd39ba6f3880c5ffc3e : SUCCESS in 14m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3d73f0c36f84a6bb5248a4d4ac12f7f : SUCCESS in 46m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/005eb89e7a654b55b515198b56ea6761 : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":1},{"id":"60e9c0cca71e3088e08bf4398fb47a2aff03acb7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 17:35:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd87612d4f243ab2553ec1546ed0d6c666b0a7d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 17:37:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e9bedae6b1448acb8d2ab292fb5ebf4 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d51600580e73bab9c9ff0c4db27d56356d5571a","revisions":{"5d51600580e73bab9c9ff0c4db27d56356d5571a":{"kind":"REWORK","_number":1,"created":"2021-09-01 08:41:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/806872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/806872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/806872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/806872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/806872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/806872/1"}}},"commit":{"parents":[{"commit":"11e162f5cc307ff77802755a8b6fda8cbdf9d4ee","subject":"Merge \"update constraint for python-ironicclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11e162f5cc307ff77802755a8b6fda8cbdf9d4ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 08:41:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 08:41:56.000000000","tz":0},"subject":"update constraint for python-zunclient to new release 4.3.0","message":"update constraint for python-zunclient to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ibc2e2c947122a4914e75ad8e078c65cdce0daec5\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+1: hongbin \u003chongbin034@gmail.com\u003e\nmeta: release:Code-Review+1: feng.shengqin \u003cfeng.shengqin@zte.com.cn\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Iaf000ea45987d29f5d12a3118d8192a9f6511001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d51600580e73bab9c9ff0c4db27d56356d5571a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d51600580e73bab9c9ff0c4db27d56356d5571a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
