)]}'
{"id":"openstack%2Frequirements~806934","triplet_id":"openstack%2Frequirements~master~I06b259e94a735b11e0d1ab3170857c07fb3bf0bd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I06b259e94a735b11e0d1ab3170857c07fb3bf0bd","subject":"update constraint for python-troveclient to new release 7.1.1","status":"MERGED","created":"2021-09-01 13:47:20.000000000","updated":"2021-09-01 17:38:13.000000000","submitted":"2021-09-01 17:35:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806934-new-release","meta_rev_id":"3bfc1e7b1267ab955acd8feb58762c9866382c0a","_number":806934,"virtual_id_number":806934,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 17:35:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-01 15:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-01 15:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-01 15:32:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-01 14:09:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-01 15:17:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-01 15:32:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-09-01 15:34:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"74198a73fbd67d36dd3b6c79a20b4bb89830b997","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-01 13:47:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77543260079bebe9e4fdee3616d9f97280a61b51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 14:09:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4f43fbe612eb45a6ba52cc2e15ccea2b : FAILURE in 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"30ef036519b3168852ec3616c18b75c00d7256a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 15:17:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b902164d30412b9fab2b1de4e106cb : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b5c13a8c47a4bde8675e5c283888d5c : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b47796bb6ae3487cb1ac76853a1e23ce : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49d79f110baa49fe9d82ce5640a77ff3 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/650c7ada57134199805957cb1a73ceda : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8906113be6534c169baa59cd5feea325 : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fe2045df0894be3be8be1dbdbadf290 : SUCCESS in 5m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d1a3891ac8084273a7ec8cd3f360d51e : SUCCESS in 11m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4718552451ed43a6839e776d70b2e1ca : SUCCESS in 11m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5aa2fd51d7e84c2db3fa394fc66ef771 : SUCCESS in 8m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1890742755114fdcb31e00f8d2df905a : SUCCESS in 9m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9b298566b0e5496ebd3112899afa4d61 : SUCCESS in 6m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8f9eaf16861e47668efa99197635f818 : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36e32ff4efaa4e7d935afabddf9eabfd : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0195701cd929420fadbef805fd51552c : SUCCESS in 10m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/eb9a69b98cc64090b0b6a632adc08823 : SUCCESS in 20m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/aaefe2243c04420d89961bbe9154d637 : SUCCESS in 12m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b3ac73630ca4414fae684fd12f69a7f9 : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2ee5d56bac9e4c688491e276e053a9d8 : SUCCESS in 23m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/82394d7416394ce5b9df9fa734d598f3 : SUCCESS in 15m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/871fec254c2442bba8b69a4b8c9a9ef2 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e99fcb9fe496407aa6f5c1b1bcf1c5c7 : SUCCESS in 19m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/745d347fc0b346f28567ab31ed5a659b : SUCCESS in 7m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/64b7f306111c42fd8d1cc4446122213d : SUCCESS in 10m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9f606064776e43ae8dc77eafd46889b6 : SUCCESS in 17m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/257468f08b1c480d84d4596134ff72d8 : SUCCESS in 43m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c588b7049ff4294a3656ef7a6b1dc73 : SUCCESS in 1h 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ada05b5a26808d52e1f82ddb3878cb3e288f2700","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-01 15:32:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e61ad09715966e3a643e7f8d57260be49900779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 15:33:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12090873df7fa9de69d57efae87f0ffc605c3ca2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 15:34:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04a66425db93b5a2ec5f0ee1295571b670a99f70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 17:35:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163971508d3745108cfa77b9d7342c05 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389bd16969f74badac71e37f9d9d094c : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f83a992320754048b5b10354635f360a : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c24e82b6cc8466c9f014570a35a6a1a : SUCCESS in 7m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54f94d67248e4ac69677f3d761a9079e : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4477132889ef41cc8aaa35d713141076 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7418e55c8f174b4b8a2ff79851b5b004 : SUCCESS in 11m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/952a1569a87f4306ba5e5da6758f5938 : SUCCESS in 11m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cf78af7aa9384688947f78d3cfe87c09 : SUCCESS in 9m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8d5c6f084b604ec3ad861400ea6eaaec : SUCCESS in 8m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2995f3db23404b008bed02c6b138a569 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5a7ab47ce0a4091b22c8dcbb353bcbb : SUCCESS in 10m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f11131f1e8ca46c4a99efb770b532f92 : SUCCESS in 8m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/478c0861be194cdb9d3b81060b710754 : SUCCESS in 16m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/beb64c3016be4d779483a88cb6b2fbc3 : SUCCESS in 10m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6875ebf7178242bd952a52696f0c6072 : SUCCESS in 7m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2a18aa1c0c8f482d9fed46962c45dfe7 : SUCCESS in 29m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5105bc9194fb4c7b8acc487b50aa2d9d : SUCCESS in 17m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e4ac243ecfd64ae5a6fa4f72c3ccb522 : SUCCESS in 5m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d0eb867a0a9744d2b69910f4642beb7b : SUCCESS in 22m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b9271d0664b94355a858bc3b5dab6bf0 : SUCCESS in 8m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/384bfc1146d3444bac471adc3e17c11a : SUCCESS in 10m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/52d91f700c0e46228157247afcaff30c : SUCCESS in 15m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb8ced2ec43b498e92a06f2c076d0fba : SUCCESS in 47m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57cf875c5ec349eda88f76bb7380d6de : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e4e64d1efc655d126563fdb664829091f4b99515","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 17:35:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3bfc1e7b1267ab955acd8feb58762c9866382c0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 17:38:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8526dde552704fc1a8aa7a986dabbddf : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a7cec25243827b9a4a5ecb83b9bc897675500da","revisions":{"2a7cec25243827b9a4a5ecb83b9bc897675500da":{"kind":"REWORK","_number":1,"created":"2021-09-01 13:47:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/806934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/806934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/806934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/806934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/806934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/806934/1"}}},"commit":{"parents":[{"commit":"11e162f5cc307ff77802755a8b6fda8cbdf9d4ee","subject":"Merge \"update constraint for python-ironicclient to new release 4.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11e162f5cc307ff77802755a8b6fda8cbdf9d4ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 13:47:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 13:47:19.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 7.1.1","message":"update constraint for python-troveclient to new release 7.1.1\n\nmeta: version: 7.1.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic98587141f737b65a2bd57be50b08a1c8b6bdd44\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I06b259e94a735b11e0d1ab3170857c07fb3bf0bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a7cec25243827b9a4a5ecb83b9bc897675500da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2a7cec25243827b9a4a5ecb83b9bc897675500da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
