)]}'
{"id":"openstack%2Frequirements~806966","triplet_id":"openstack%2Frequirements~master~I9a06ca1773b690a5eef51e9fff90ecca7f968167","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9a06ca1773b690a5eef51e9fff90ecca7f968167","subject":"update constraint for python-glanceclient to new release 3.5.0","status":"MERGED","created":"2021-09-01 18:28:07.000000000","updated":"2021-09-02 01:33:29.000000000","submitted":"2021-09-02 01:31:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806966-new-release","meta_rev_id":"4ccdc18eae6bee73d9b1e6d0b42722391a830c23","_number":806966,"virtual_id_number":806966,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 01:31:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-01 20:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-01 18:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-01 20:16:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-01 18:51:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-01 18:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-01 20:04:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-01 20:16:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9aeb47878ccb3cb151adf4e97c23afc93a94cb03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-01 18:28:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ad4cf2a12b8c63d0a41874c863c4dc73aed6d25","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 18:51:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ecdf621fc74d4accb74b22cfe554bbc3172bff3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 18:54:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f6d84068589047a688a9a58873da2499 : FAILURE in 11m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d62a18c3a07967eee027b448545bee004d7f76b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 20:04:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec606b54da934ea39273ec8051893746 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177a577b64354b408ed598218551d5ff : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1962b1d8fa6f4123bee4337ae5c51286 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/099064fe547e466ba63ade2c1072edd8 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f868c7daf06b49ad809a5898c42a051d : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4af7d51c5b26499ea544d5ab287ec796 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60b7d9be64544d2299be0c4ddcc0a82d : SUCCESS in 6m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2724e2da7317479c836d993149dad969 : SUCCESS in 10m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2a968ced776b40d09a9852e16ae76482 : SUCCESS in 10m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8c7c38c3485b4128969639c9fe38f20f : SUCCESS in 5m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d1d25edeba5045f58b7263df243d9145 : SUCCESS in 12m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ff2e6e4a89144dfaa3b51e32496f863c : SUCCESS in 5m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/81c0d60bfc6e4b84894b67a1aba4d21f : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f95370e5c50f4ac58f20ddb33104b8d3 : SUCCESS in 10m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/68a86b99351d4382aadff49386285ba5 : SUCCESS in 10m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fd415d249bde4cebb5fb51b6133d6f91 : SUCCESS in 16m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/846b8dd6fa124762ae357ddc9c5cd88a : SUCCESS in 10m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2080553b965246b19cf944b3bb296349 : SUCCESS in 8m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/32fe0411832d44b5be7328e10d7d94ac : SUCCESS in 24m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2378d5ae9a3c4cfba101b17a12dedf63 : SUCCESS in 13m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fbc37ef8598b42fc9a5fdf42b56f509a : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/842fb47fb32f4de4a59e3c76404d2e10 : SUCCESS in 19m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a8a93789b1134f68be29f8d21ac0001e : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91fff25dcc964435b3425e96821092ad : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b29842c4368f47ba96e7e09b8e828786 : SUCCESS in 15m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93e0e5f221244635a9ed4ff97ac87137 : SUCCESS in 40m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/724562fc89a44fc7a54976403a9eec6f : SUCCESS in 1h 30m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a8d3d8da0b04bbdb4b0797ac0c2d1d1bae5c142a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-01 20:16:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"234e6c31bf2db99b306e5f203d10253b730e33f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 20:17:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfdc62e4bd077b5ebfe22072fd06e4ea2bdb960d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 22:44:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/620bfb61bfae4a4b869cf2ac60fbd75b : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a29a2014337842dcb1311840fe388d1b : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/173416416c7d402fa217849c3ef93c00 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec63e9c405ff4e57b47206f585e74028 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d130bf221df4c07b73d16de5d6e284c : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15b3afd101f14984948e5948d54f72cb : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/483b49fcdcb24cbe98bd5813ba3ef7df : SUCCESS in 16m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db32c5b2c4ea435f8caeab86712723be : SUCCESS in 9m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/522ee498994949dd83a22b5b376d0e8b : SUCCESS in 13m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/66ab2879c35943bf995e25d3fc089ff5 : SUCCESS in 7m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2069ec0753e54974a39cea3d096963a3 : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5724433c52d242c9af0f9e80de29549b : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d89026386e1d41f0909d0069d27648f0 : SUCCESS in 9m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5eeb68f55c75487897f9f4e67dc441e8 : SUCCESS in 16m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c8328d2f9bc34a2d8a145c2ed60fe0ed : SUCCESS in 10m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f6ee82fad9074b979ce51262651bec92 : SUCCESS in 5m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9e91d8eca7094d7cb4a271b47ddbdb9a : SUCCESS in 21m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7c66c015819e4a67ac94e71d4fcb7706 : SUCCESS in 15m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/772645f116454824b486b6153b810f22 : SUCCESS in 6m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5ba3765b42a84592a76aea1fafb142e2 : SUCCESS in 19m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4c1896ba8fbb4827abae3ef9d808b1f7 : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b1664991a89645f1b6c368700a90bc00 : SUCCESS in 8m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/84f38016c066456886b26af5e21eab7a : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1a1d6f4ccdd4eefa5f876584f3026e1 : SUCCESS in 43m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1b256b6656a4496939a0b69a46dd1cc : FAILURE in 1h 00m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0a5d24a063b40fda87d894cc05ead5aaaf9cd706","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-01 22:49:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0001fdb91c703f12a73f9a5908681b44590ae9ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 23:09:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a33bd239791744b4943fdcb3e3267f35 : FAILURE in 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0fe7bd181da9d3c44f03f25fa4ead84aca7bbdae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 00:05:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f384aa0021804e9da1664e17ed8a8235 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b0f8ea581a4611954137d86f2bf06b : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90306f3c0d74cc980e0678b1b2b5167 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4316048a2c8d425791c0fa3c5026260a : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59a150f3d26c40c7a04148302e274a03 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88d1e47597384e75ba55b6bf500ee400 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cd26b1bb75b34da78a03a83b7e65705e : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/240d401912ad4a33ad07396339188c43 : SUCCESS in 12m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db4e7e05f88a4ef88848872dd13a4db7 : SUCCESS in 10m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/044f99ba71844afe96d5dff0da258e73 : SUCCESS in 6m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/df0b94d2302a4656aaf4f1223497aa25 : SUCCESS in 10m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e4ec38a68d7e46d6b6f9b9c515c7d7e5 : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/176051fa1b5f42399fb9695fd043d2f5 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e09c207f0dae4939b8a727030caacc70 : SUCCESS in 10m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b652fcb9e37041828e1fe2737930c5f6 : SUCCESS in 9m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/afd19b4554c9493f8a855cb7a0dfd01c : SUCCESS in 13m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0aa6dc180b2548cca74d4e57872b3cc8 : SUCCESS in 9m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/354f5abb5338450d982b66b103b5b0d1 : SUCCESS in 10m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/520e8e5877f74d3ea05a0bc7d772fc2a : SUCCESS in 21m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/83a9ac4374344143b2afe47429916bf4 : SUCCESS in 14m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c04ff764b8ed4aaf93a9771d0a8e4deb : SUCCESS in 5m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0ba35c3c63b4d6588a209248175cc63 : SUCCESS in 17m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c26b25c24b5c4bf6aae0292496931cd6 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dc923809fa974dcc8e4ecfb93d42382a : SUCCESS in 8m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e9803ca249594f5c824da30c36f14861 : SUCCESS in 12m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b387d8ed28404abe827f3b73bf2f7eb7 : SUCCESS in 41m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2d6d79363f947a988adfd0e420d8a6e : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7e026d8c42776f5117e86f3498bb8f592df4d87f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 00:06:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c6b6c62d25f346dd58af6e31fa8e553f613df10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 01:31:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd69f389290f4ba49b4c9558ece16368 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc2ce6a20704effa27f1c7bf6e44b95 : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a46fc80f0e24ab483077ff23b2df174 : SUCCESS in 10m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b12e02dd4174d3ab6e9551a6bed8ce8 : SUCCESS in 10m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce9032a9a69f4ec09b51a1b7a2fec2df : SUCCESS in 11m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35b5120b7cb34a0d87c2d56d32fe0494 : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/95e9969b2099467c9610a31624e2d1dc : SUCCESS in 15m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ca598a0e83b64aa287363523e999a267 : SUCCESS in 19m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a0aff12eb12049d18f74e9c1706fb1a8 : SUCCESS in 9m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6d894809838449268a56973dbcb55bcc : SUCCESS in 10m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7d281a3679cd4590b667e6151b19fb20 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a65629d49bea4074bd5565a7ad15f471 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ba739556a18407886e477de70fe4ba8 : SUCCESS in 13m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3fb750302e54408e96ec6b103ed2fd64 : SUCCESS in 13m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/64a90c9e08b94185a3f92c83c6d5e1e6 : SUCCESS in 11m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c0c8e09953c044b7ba540ec130914f35 : SUCCESS in 12m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/06ad83235cea49f8b02a7ee330fda402 : SUCCESS in 21m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2cad47e898374cfc95da081c46afb119 : SUCCESS in 14m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e35e61932d54c9f9bcb84126927971b : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06e9267b575143e59766beabd8cf639c : SUCCESS in 18m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/43b56a266c9e4041b89855192618eb15 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fee957a8c9944e8da31266435bf040c8 : SUCCESS in 11m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a4349afcebb941fe87afe7c9a28504af : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab18a4018ead4b4a889e1226d9fbc387 : SUCCESS in 37m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8ff8cd846064194a04cbe5c6697f192 : SUCCESS in 1h 23m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8e13eb0fb542c39a71a78d8d08f1270e762bcace","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 01:31:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4ccdc18eae6bee73d9b1e6d0b42722391a830c23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 01:33:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff71df3e4e5c498b8f0a5b99bbe6af90 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"185b1544661f393f105dcb03bc38a991eab46f64","revisions":{"185b1544661f393f105dcb03bc38a991eab46f64":{"kind":"REWORK","_number":1,"created":"2021-09-01 18:28:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/806966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/806966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/806966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/806966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/806966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/806966/1"}}},"commit":{"parents":[{"commit":"f1ca47e8d22e5387d2ade7b63ebfe4421790b944","subject":"Merge \"update constraint for python-troveclient to new release 7.1.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1ca47e8d22e5387d2ade7b63ebfe4421790b944"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 18:28:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 18:28:04.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 3.5.0","message":"update constraint for python-glanceclient to new release 3.5.0\n\nmeta: version: 3.5.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I41a47fac39a275db84a1d97d333013c89777a935\nmeta: release:Code-Review+1: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I9a06ca1773b690a5eef51e9fff90ecca7f968167\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/185b1544661f393f105dcb03bc38a991eab46f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/185b1544661f393f105dcb03bc38a991eab46f64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
