)]}'
{"id":"openstack%2Frequirements~806969","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I36ba6aaa3bef25a22dfe12a033dacdf2a20687d5","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I36ba6aaa3bef25a22dfe12a033dacdf2a20687d5","subject":"update constraint for os-brick to new release 4.3.2","status":"MERGED","created":"2021-09-01 18:33:29.000000000","updated":"2021-09-01 23:58:58.000000000","submitted":"2021-09-01 23:58:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806969-new-release","meta_rev_id":"b7491af999dfd4cd393c7824231e18e0cbbca4db","_number":806969,"virtual_id_number":806969,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 23:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-01 20:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-01 18:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-01 20:16:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-01 18:50:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-01 18:51:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-01 20:03:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-01 20:16:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a9f012197ea479f161160b610f62fac82dbc3a5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-01 18:33:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a5db918e0c99da72da5fae02cde26e472838f79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 18:50:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8c9cced5a2b2460bb7b7f2a21be51ec1 : FAILURE in 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d071c2af890935e3d58ecc806c33c8b4f32bd898","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 18:51:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acfa3516724f18de8ecdb9c27dc2154a1b70d5c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 20:03:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0fce181c52c40829bdca082d4137f7b : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d94a160605bb4acd95fc4566c857f638 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ecd931a06664bfcbe529b815abaed0a : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7adb5c924f14d748ed5bed52983c7a2 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68c609bcfe7541a49f83ccf0d35a89e8 : SUCCESS in 6m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/500debe588da44c6bb965eedfcf91654 : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/817ce01c10fc4409926c6cc4a605cda8 : SUCCESS in 12m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73c0b98b3739493793b058645cf000b9 : SUCCESS in 11m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e44c79a3ecb044219c989f8e2fc4aaac : SUCCESS in 11m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ae5f859faa984c078fb71db676abaac1 : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f5d5674bc47c41edbc915854f276c619 : SUCCESS in 12m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/610c7551a205499c8c43d2d75970d571 : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b09cd8d3d9d94c63b396435d83e242bb : SUCCESS in 10m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a2cc936f1e6b4e3fb54878b687aacc39 : SUCCESS in 21m 26s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a9cd0325f7bc4a5f96df3f590305f89b : SUCCESS in 5m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e8f05d39c28c4bdbac950533062e6c92 : SUCCESS in 29m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/045424ec5973484c8ba617b7f700472c : SUCCESS in 13m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d1f9ce75bf647b89436200d40a8b67c : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7d1c6f9bd4248ff9a7ab6c22a63b61a : SUCCESS in 19m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e79615372466438a9fbbc717421c8091 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b447cef54f641aa9b83d38251e996e4 : SUCCESS in 6m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/80dfb1bb318c4bf88f8a92e506b4d662 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/131385d5182c4e7d8a40bb60eb4e642e : SUCCESS in 53m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8170c326e4c84e31bd8008662b042794 : SUCCESS in 1h 27m 44s","accounts_in_message":[],"_revision_number":1},{"id":"31e3eef506ea506898dc754472be288d8aa00563","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-01 20:16:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31a2273bb960868482a39b682573684a5339b7fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 20:17:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12bab440ab152224b56b8ebec6968acdf087e1a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 23:58:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf48bf3dc3534085b73cd5ea006a37b2 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bff9f8190ca44f79d8fa797cfe84e05 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0d94b2a834641009bdb2049e175e87a : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2d68b58a73d46e4b81d66502bf33038 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fb471c2752d4d80a5a7e90e8f863088 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fcaef45315f44969b9e928dba06637ec : SUCCESS in 9m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae74315987ac4e4396b8b52f44372368 : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bb06920853fa45aea4c972cdadb149da : SUCCESS in 9m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/508b280c95fb4d9a9c5a49493f5ef897 : SUCCESS in 6m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ebb0caf0354544d1b83e1b427037157a : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b084cf8bc7a4d3db8337dcbab22a93b : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/11d57a435fd74483b383d46d8c8a3adc : SUCCESS in 9m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f1a9d94748584040b5a3f0ea4e558099 : SUCCESS in 13m 56s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/d2ca141a1a5741f19018b2f6af6279fa : SUCCESS in 7m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1407beb0b99c4c10aabd697476ede33b : SUCCESS in 26m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/99fe6395c46e4ea394c652f1688d2673 : SUCCESS in 13m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1305f39ad7c74881acb6bf92fcf16908 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/327e7bcd91994efd8c7fe3689fac75e7 : SUCCESS in 19m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7857dc8a8bad4df08cf7b26679538dab : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9df819fe07b84a7d97695c3cd306efeb : SUCCESS in 7m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f01caec1480b4615b5ecebb1108c1ae9 : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ada3049dcb8d435ab6930b084a68fdbc : SUCCESS in 1h 01m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37abdebcd448422aa66795b8b5d25bd4 : SUCCESS in 1h 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b7491af999dfd4cd393c7824231e18e0cbbca4db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 23:58:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"952778ad1f33e3c84f4bd013c766c292d3b372e8","revisions":{"952778ad1f33e3c84f4bd013c766c292d3b372e8":{"kind":"REWORK","_number":1,"created":"2021-09-01 18:33:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/806969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/806969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/806969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/806969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/806969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/806969/1"}}},"commit":{"parents":[{"commit":"11107bc997b94c91e93bf7d9f487eb569dc4305c","subject":"Merge \"update constraint for ironic-lib to new release 4.6.2\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11107bc997b94c91e93bf7d9f487eb569dc4305c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 18:33:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 18:33:25.000000000","tz":0},"subject":"update constraint for os-brick to new release 4.3.2","message":"update constraint for os-brick to new release 4.3.2\n\nmeta: version: 4.3.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: whoami-rajat \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: Icbd82c6841d443d8f008f29e7ef95bd207da3e20\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I36ba6aaa3bef25a22dfe12a033dacdf2a20687d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/952778ad1f33e3c84f4bd013c766c292d3b372e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/952778ad1f33e3c84f4bd013c766c292d3b372e8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
