)]}'
{"id":"openstack%2Frequirements~806970","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I9eeaa63b86dcfcc719a34573d633fbd8e65e00f0","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I9eeaa63b86dcfcc719a34573d633fbd8e65e00f0","subject":"update constraint for python-cinderclient to new release 7.4.1","status":"MERGED","created":"2021-09-01 18:34:28.000000000","updated":"2021-09-02 00:16:36.000000000","submitted":"2021-09-02 00:16:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806970-new-release","meta_rev_id":"2a196d0d36a75b02665ce4a5ffaceca21c5e2d0d","_number":806970,"virtual_id_number":806970,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 00:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-01 20:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-01 18:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-01 20:16:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-01 18:52:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-01 18:52:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-01 20:11:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-01 20:16:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"054b3fbfecc2e81237aa35d1316ffc852a59eaf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-01 18:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02f6da7f5479a91cc26ff6a4a21d7e10c717e137","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 18:52:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d133675a155cbf8e7103abfae35cc7a30853eb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 18:52:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1988678b635048babfd26a2834632753 : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a64664a54b31df52fb487271cdcf1af74469658a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 20:11:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f63056fb9f9749ecbaf665322f663376 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f09905dacfb3468da990b72aff81bee6 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cba06a09dcac47d8a84b3b3b05d220e2 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56416ba9716946aeb064c3fca2f896fe : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a62ade5dca8a432a81bcdf249d239e81 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/54d787766f6145dc8fcde958db75a20d : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b1b00f71a2d544dcb13597962a8a10f7 : SUCCESS in 11m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e526b475e4994f1097fb09f5cfad23a9 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/94950f3be9004314afb8f5d5507e2e74 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1c68c51ee15b45dfb67e947cefa4ba53 : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b4d2a7ed6b914c768f57de6f2bf041cb : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2ba038689984f27a223ffce207f8939 : SUCCESS in 11m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a24efbd453114e07b6e4330b4d51ca49 : SUCCESS in 8m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f9d37effda249c68cda7540e5ea978e : SUCCESS in 12m 27s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3b75b644d7fc49a2a9d876a08d94dca0 : SUCCESS in 6m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/efabe56aba9d48278d6629601bfdbc45 : SUCCESS in 31m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3162ea697d144025a9c3f6c83f872b52 : SUCCESS in 12m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2624a34ffc164ba99ef041c1c74f4c5c : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/911aa79bef324c7c88257c635d728b7d : SUCCESS in 21m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c431a9f44a0e49d999f127abbb3d4f71 : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cdb050478e0144fdaf381c905de8e12a : SUCCESS in 7m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f7966bceece942108e3707975d3a9739 : SUCCESS in 15m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff8d42e4b7844b06beea9fec1d7bdd13 : SUCCESS in 54m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aefd1aef77c04f348bdd44a7e00d80bc : SUCCESS in 1h 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2c72063541bb45694a0f06aea5f24d353566f1a9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-01 20:16:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4eb693dbfda9373835ab03501b6871b0db9ae3ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 20:17:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30d25a916e9bbc6dcda6e6df6c8fa81a79fdf8e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 00:16:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18d2425d2354376ad4efe24e466c01f : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e50e516b12b64356848af15563372d7d : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c11461f95f2444f28dd908573d80d506 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b914dfac21b2414b84f5a95968c1e535 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/482eb52e2ef949d295e7177e12d20de2 : SUCCESS in 5m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2b09db27274942d2a6bbbe49fd53f21a : SUCCESS in 10m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3be79c94722a4e879ccaff1406e2c334 : SUCCESS in 10m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b3609f7e326b46188280e96d825650ef : SUCCESS in 10m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/263d448a51794b0198f859b7ee972b58 : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ca2a04507b94460899775b21702eb602 : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c1653e19de034a109fdec56359351161 : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7dc2574897db46cc9b3296283098ff1d : SUCCESS in 8m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/24d7784b3cfb4be381698efb246c177e : SUCCESS in 13m 45s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/732636544f174b2e8479cc900077ca61 : SUCCESS in 4m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6dbf327f7a7640788beef43bb226b48c : SUCCESS in 31m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a45ab0351b7246939e4768bd100a76f6 : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c8cf890e3a744938afa88cbd6e07d8e : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dc97664101934635a5ec624afcff1daa : SUCCESS in 18m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/12e31096378244dfbcd36552ab714219 : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a120d6a1b4174ccb85e6815772daa985 : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1a045aee92c24ab597f95d11a4cd9d38 : SUCCESS in 12m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cb19deb9c9d4fab833c41bc2b26b181 : SUCCESS in 53m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb1954cc6d60443cb71da1c625edff71 : SUCCESS in 1h 42m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2a196d0d36a75b02665ce4a5ffaceca21c5e2d0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 00:16:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c90120aa7f807eda17a119d31e8b84004a99f81","revisions":{"4c90120aa7f807eda17a119d31e8b84004a99f81":{"kind":"REWORK","_number":1,"created":"2021-09-01 18:34:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/806970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/806970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/806970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/806970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/806970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/806970/1"}}},"commit":{"parents":[{"commit":"11107bc997b94c91e93bf7d9f487eb569dc4305c","subject":"Merge \"update constraint for ironic-lib to new release 4.6.2\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11107bc997b94c91e93bf7d9f487eb569dc4305c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 18:34:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 18:34:27.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 7.4.1","message":"update constraint for python-cinderclient to new release 7.4.1\n\nmeta: version: 7.4.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: whoami-rajat \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: Icbd82c6841d443d8f008f29e7ef95bd207da3e20\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I9eeaa63b86dcfcc719a34573d633fbd8e65e00f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c90120aa7f807eda17a119d31e8b84004a99f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4c90120aa7f807eda17a119d31e8b84004a99f81"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
