)]}'
{"id":"openstack%2Frequirements~806971","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I0c7909d993d9867d8167f38da501071fd0192d3c","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I0c7909d993d9867d8167f38da501071fd0192d3c","subject":"update constraint for python-openstackclient to new release 5.5.1","status":"MERGED","created":"2021-09-01 19:10:51.000000000","updated":"2021-09-02 00:16:38.000000000","submitted":"2021-09-02 00:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806971-new-release","meta_rev_id":"275ca78f6bf60f2b4e706ec7a980485283f0ef03","_number":806971,"virtual_id_number":806971,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 00:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-01 20:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-01 19:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":2,"date":"2021-09-01 19:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-01 20:41:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-01 19:11:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-01 19:21:58.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2021-09-01 19:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-01 20:36:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-01 20:41:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4f01629c7a31d485eb24b580fbf09af1e71b4bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-01 19:10:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2483d9084522556ff78d0c362902e1a20b2564f1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-01 19:11:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26e6ef5ece4fb99e1f937c010512f6a347989f69","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2021-09-01 19:21:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"534df9c59725edb965bcd9bbcbaab0565f40fe7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 19:27:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8c4515bb51dd4024814031d0e8c4497f : FAILURE in 14m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d36db5f0422b0558c2477def7bca25278cf14693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 20:36:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad52949c46e44c0ba0b17e6c2a45b436 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f9324409d4044b98acee37e42263284 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e44692b1e164b6381547a76465f86b7 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/991c4f9b5a9b4ee4bc6c522b733e42a4 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fef86e7736464a848a62d5ff3e02e16e : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2297915ddf614688b8318165bc6bf5bb : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4286b533f3c94873920c50620afe5680 : SUCCESS in 9m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/df4a9aaa31504c459c8d2aee24ae27fb : SUCCESS in 10m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/47d73685ead744be999010b55aaa8e0d : SUCCESS in 10m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5f5ae152413143df974e8928a16b00c2 : SUCCESS in 5m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eeda6d77bf73422bae06954ac1e195f6 : SUCCESS in 14m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f77585e99f33479cb2edd11e89030c22 : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/71c2c4bcc38148189c5b70d3217376d1 : SUCCESS in 10m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e4406f83b5204bb29e482eed5e9138e8 : SUCCESS in 14m 12s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5b0e7c232f6c43f984db6ea08bce2ea7 : SUCCESS in 5m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d4c9dcaa4b9240779d5749c8ecc0c90b : SUCCESS in 26m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/688a9109b97344e2ac6d5c745f2ff4e4 : SUCCESS in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b858dc5c789245aeacae4d07db56a994 : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1f24772b11004673b94b6f047dd57d3e : SUCCESS in 18m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/622504f466764176b549fd315e323cd7 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ebb554cc1126466ebed3c5c38182ecc0 : SUCCESS in 6m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5087144b15c349faaa7f046fd5465828 : SUCCESS in 11m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d64c3ee7660a41b1b10287c21b781703 : SUCCESS in 54m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7717bc6a51544b22b7e379fb349b831a : SUCCESS in 1h 23m 04s","accounts_in_message":[],"_revision_number":1},{"id":"22fa7b930ec321b6d2961d53e783763058116fe9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-01 20:41:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb43b59a29de0093af35b029bd101e9836197cd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 20:42:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f73588babb94d4ef77f67089cca199d178b29c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 00:16:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0ea8f391aa41b099c1dcaeffebdd9c : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de515ec268b24e3bb8ada3d7b7f744aa : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec202f55e72a45409c3c26e26b52c90a : SUCCESS in 6m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af2e20fa10a34359bf48f4fcc8abcda3 : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d72f19d81e1448acb58b26feaf5291ed : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d614a62871d407e9b7c32818b2a04e3 : SUCCESS in 13m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/575d1b7f7e1344edb2da43740fd388df : SUCCESS in 9m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/63307dfc79ff4f6c9a40803ca8a4fcd0 : SUCCESS in 12m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2ab27309a82a4d0d87d16778342ace4e : SUCCESS in 12m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/60c14417ac7048abbcc0d6254f56586d : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6815946c2c6c42ab8c5bd08f698c2dea : SUCCESS in 18m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/26fe0dba84d244adbf816af69604db68 : SUCCESS in 10m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/020d5c49d6a045bd8a95960cd48f2d39 : SUCCESS in 25m 06s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9d87e654587745e2be7091252ad375aa : SUCCESS in 7m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c1b264d87001443e9d4f4acf40deb839 : SUCCESS in 28m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/69ac33c2523748449c869309fe3d4b96 : SUCCESS in 14m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c66101f492ca46489a944fcb52eb950f : SUCCESS in 5m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5606888adb7a4496820cfbfa7a12fd3d : SUCCESS in 19m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cff64969444f4dc49cccabddbdc3b16f : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbd060646c8b4ea0871dbadac8d07b4b : SUCCESS in 8m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/12f0d6f620ea4a558cccb5bea183d610 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a49066ecac041359b92240d90b52d78 : SUCCESS in 47m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7b347743b0f4dd08f2c187a3783732f : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":1},{"id":"275ca78f6bf60f2b4e706ec7a980485283f0ef03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 00:16:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00efe860da9f08b22dfa7ee56412f95a909e7c0c","revisions":{"00efe860da9f08b22dfa7ee56412f95a909e7c0c":{"kind":"REWORK","_number":1,"created":"2021-09-01 19:10:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/806971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/806971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/806971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/806971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/806971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/806971/1"}}},"commit":{"parents":[{"commit":"11107bc997b94c91e93bf7d9f487eb569dc4305c","subject":"Merge \"update constraint for ironic-lib to new release 4.6.2\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/11107bc997b94c91e93bf7d9f487eb569dc4305c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 19:10:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-01 19:10:50.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 5.5.1","message":"update constraint for python-openstackclient to new release 5.5.1\n\nmeta: version: 5.5.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Javier Pena \u003cjpena@redhat.com\u003e\nmeta: release:Commit: Javier Pena \u003cjpena@redhat.com\u003e\nmeta: release:Change-Id: I4372568fde58319c5d96ca0614411d96a98c562f\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0c7909d993d9867d8167f38da501071fd0192d3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00efe860da9f08b22dfa7ee56412f95a909e7c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00efe860da9f08b22dfa7ee56412f95a909e7c0c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
