)]}'
{"id":"openstack%2Frequirements~807237","triplet_id":"openstack%2Frequirements~master~Ic47c58a39e550e95d535466812adf7a192a8185b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ic47c58a39e550e95d535466812adf7a192a8185b","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-03 06:37:18.000000000","updated":"2021-09-03 21:22:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f8dbd07d1933f3c5140884ac9975e3d4b256c7e2","_number":807237,"virtual_id_number":807237,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-03 08:34:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 06:54:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-03 08:34:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04a6986441fb513d129fb9e7527c1ed434e65428","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-03 06:37:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da4306d92ebd4a793894ced4a3566ce9b30c30ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 06:54:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a701d5be726043ba87b25607b6f3a802 : FAILURE in 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c3335d43ca0478cfb59408d0d1f87fd745dde362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 08:34:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e40ca8869ef8445e9f8259753ae1dd9e : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b68d5a3d6d44559300a22c2e628594 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c3745a0f1ec471b8789d0541a63f374 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45c1677de7234255ae7525d7865d0090 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/832ec4f06341481b873d1f1577469dba : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d8e5046ad894fe59652495a3ed34227 : FAILURE in 5m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb0df6b640214156966f4d8da87aa4cd : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15288620b7164f6abc663432772316d8 : FAILURE in 6m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b5a1aa98c3b408dbc4720672c715bb4 : FAILURE in 6m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4df8af5b6ef84d709f2203f9a955a6e6 : SUCCESS in 7m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c693826ddc71454ab63b0d298bae5288 : SUCCESS in 11m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c716e047264f4608814a9baf72e32188 : SUCCESS in 7m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a88f9a1d06b48ba9ef976fb552423ea : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03e8d6b86ced434899c0c6455a2712fa : FAILURE in 11m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1c49d6012a624e779e1a02484080a945 : FAILURE in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0d67a90ce353465298a15227e476792e : SUCCESS in 15m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5f44040f1c1f4986b8e2052a445ec593 : SUCCESS in 10m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7efb6435cf674307865dd99511f49522 : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/530b94a79447470f97b5f089f3126bee : SUCCESS in 29m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/41eb07cf727043709bd7041a3e8d3dc7 : FAILURE in 15m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/603f7dd640554551a4cdb0065a5fd72b : SUCCESS in 5m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30577d8f8b794ee69ac854a6996e32a4 : SUCCESS in 20m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/78443efead8b471b8cb35aedd0cc300f : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b26b27adefb412abb34449721b0ff41 : SUCCESS in 8m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/44b8e3de8c564a82b7731bb9b1f26fd7 : FAILURE in 15m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80a35637f4754ed186a8e13b1df01304 : SUCCESS in 44m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07c8400c62f94e74b25728488440cb4c : SUCCESS in 1h 20m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f8dbd07d1933f3c5140884ac9975e3d4b256c7e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-03 21:22:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91482cbc66f26c052696be8748c65a67cef60c3f","revisions":{"91482cbc66f26c052696be8748c65a67cef60c3f":{"kind":"REWORK","_number":1,"created":"2021-09-03 06:37:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/807237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/807237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/807237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/807237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/807237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/807237/1"}}},"commit":{"parents":[{"commit":"6d19f0f7de2869f014ff3f23b0e5d1000936680d","subject":"Merge \"bump dogpile.cache to 1.1.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d19f0f7de2869f014ff3f23b0e5d1000936680d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-03 06:37:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-03 06:37:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic47c58a39e550e95d535466812adf7a192a8185b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91482cbc66f26c052696be8748c65a67cef60c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91482cbc66f26c052696be8748c65a67cef60c3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
