)]}'
{"id":"openstack%2Frequirements~807288","triplet_id":"openstack%2Frequirements~master~I6870aec30af2463dfcd81ad7afc2999401bf3064","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6870aec30af2463dfcd81ad7afc2999401bf3064","subject":"update constraint for python-neutronclient to new release 7.6.0","status":"MERGED","created":"2021-09-03 10:45:23.000000000","updated":"2021-09-03 15:58:09.000000000","submitted":"2021-09-03 15:53:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807288-new-release","meta_rev_id":"3f2dacf87b2a8012279eb56341fac3ca3615c32b","_number":807288,"virtual_id_number":807288,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 15:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-09-03 13:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-09-03 11:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-09-03 13:41:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 11:01:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-03 11:16:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-03 12:05:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-03 13:41:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"0be7c7384e3b497431becb82f641ef5cea3041bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-03 10:45:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4dddf4d240927515ca94517c48c838a84f6ffc2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:01:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/48d6e709f3bd48b5901c44a3e17c3134 : FAILURE in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"befd268f47d998b32200d2758a0a33ca99daac87","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 11:16:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"284de029e34ef00dd4fdbd64d4ea49687ef6a271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 12:05:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/189e5562fd42472ebe1777b59509407b : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/590496bc8a264fc196577c263b6979c8 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dc01b7dab9341cdbb0837c461519e0b : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfa651b4a3c844d0b89ddc6959948b0d : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0323cec68f6458fb148e87bbaf9c23a : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c35d2f5d76104710b824e5c4b1b9c306 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc8f2a97f31245d28c790be6711cea7e : SUCCESS in 5m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fea5c10d45a444feba19d74e8f3d1f01 : SUCCESS in 11m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5c301a73aaed47859fdbad11a9768263 : SUCCESS in 9m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8ce660bb8a3f4bb08b0dd5c41a933bb1 : SUCCESS in 5m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e68ce34497da4e7688665580542eb078 : SUCCESS in 9m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c6dc6761f4fa41c88401285f466d9be3 : SUCCESS in 5m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e335894f29a844ac8a79044805d8b513 : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb78e9db6d7648f09385c8b3afde669b : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1460a96585ba478f974bf6c2ad377fb5 : SUCCESS in 8m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a437c4e8477d44199be16bdfd5a9d3b6 : SUCCESS in 15m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0899956698bc4e6e91e74ab2d4b5e0f5 : SUCCESS in 9m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d917738a5b84449c85f2ece8a0d0bd40 : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dce0ba75847b4cada7389fcb9797a7fc : SUCCESS in 26m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6629eaa997254880b9238eabe40f1b1d : SUCCESS in 15m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c7ae90713b7445f985731d662031e4b : SUCCESS in 6m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5aa2fe6a675149a9b1fb862e0ba00c5b : SUCCESS in 17m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0d72f46392f2416d9905a3aa1fb802a5 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66bb610bbc16442bb177ceab2cf859fe : SUCCESS in 8m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fd120c6c52c941c290b8ded752ecad9d : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a50513fae11743c5874eeefa571b43db : SUCCESS in 44m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6865de3bfa8459d8c595d5f187989a5 : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a1f75bec8d5f75a89fb9d09ee48a37c72f54df0b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-03 13:41:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc562cc8074607e7627d5515c7566fea01b39796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 13:42:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ca3bf1823590cbe2ddc0f945c834514f6640ec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 15:53:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c6149c4037147549de5da324e9bdd5e : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/441235830d4b400da88a892ab9cca070 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9d4591ee02a4de1bd2fc5947a14bcc1 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf248dad68f44cf19c0b6f0d273111c1 : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ba09c77038d4900bd6968585a9f4519 : SUCCESS in 6m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3a898a8e8b541f48c5f72e2b768935d : SUCCESS in 6m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d76ff8f7cbab4399bb6688623cf0d797 : SUCCESS in 13m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6029254373494d61a92c90eed10a2a8f : SUCCESS in 12m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c2cdfcc373ea41b885ab7b9a894a8ff5 : SUCCESS in 10m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7d1cdc7bc10b42619bb8fac91a8ed58a : SUCCESS in 9m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/747b2eda95144152854b0cc0dc1f7869 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8d068269220042369a82cf0191565cf4 : SUCCESS in 11m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5f6b035145974023be385a79c38966df : SUCCESS in 11m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bccc7da15f554195a46cbe650caa97e5 : SUCCESS in 18m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c7f009145ec24d9a88db2c761e80e5b2 : SUCCESS in 11m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/62db8ecf8e504d57ad2ef6c4a52e80e9 : SUCCESS in 8m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/369238f47efa4298a066eae9a065b99d : SUCCESS in 31m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4cf00226c80b4701883937599f073fed : SUCCESS in 16m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cbda1c481afc4fb8a4acc9d3d620f79c : SUCCESS in 6m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e97230e16755444f9d522fb219d0f2d2 : SUCCESS in 19m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cbfffb5566b247a6ad91e0a8fe4826b3 : SUCCESS in 7m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad85cd1be3b54bb3b40811ceaabee2dc : SUCCESS in 10m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d992fbc43db346209e489d3cb6cea7b1 : SUCCESS in 16m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f90154425aa4b0b9fb0fcf8ab82f31a : SUCCESS in 47m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdfc07a404a648e683f43f23ff4c8bed : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c4c4600f05ab4280f5d3da4bee2e7303142e7fd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 15:53:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3f2dacf87b2a8012279eb56341fac3ca3615c32b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 15:58:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64eeadccb37a4594a620926529d5204e : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ebe82921d23c5f61526b0a0cc0bae4a2dc702cf","revisions":{"8ebe82921d23c5f61526b0a0cc0bae4a2dc702cf":{"kind":"REWORK","_number":1,"created":"2021-09-03 10:45:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/807288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/807288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/807288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/807288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/807288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/807288/1"}}},"commit":{"parents":[{"commit":"6d19f0f7de2869f014ff3f23b0e5d1000936680d","subject":"Merge \"bump dogpile.cache to 1.1.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d19f0f7de2869f014ff3f23b0e5d1000936680d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-03 10:45:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-03 10:45:22.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 7.6.0","message":"update constraint for python-neutronclient to new release 7.6.0\n\nmeta: version: 7.6.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iaded0070ca9769d4428eecf285ae975a6c2c1a6d\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6870aec30af2463dfcd81ad7afc2999401bf3064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ebe82921d23c5f61526b0a0cc0bae4a2dc702cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ebe82921d23c5f61526b0a0cc0bae4a2dc702cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
