)]}'
{"id":"openstack%2Frequirements~807342","triplet_id":"openstack%2Frequirements~master~I2c86bce636a2aafc00934039d2a19a2365865c23","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2c86bce636a2aafc00934039d2a19a2365865c23","subject":"update constraint for python-mistralclient to new release 4.3.0","status":"MERGED","created":"2021-09-03 14:38:59.000000000","updated":"2021-09-03 21:06:21.000000000","submitted":"2021-09-03 21:04:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807342-new-release","meta_rev_id":"71482458af78198f5fa03d55db14ab5000e41769","_number":807342,"virtual_id_number":807342,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 21:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-03 16:03:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-03 14:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-03 16:03:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 14:40:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-03 14:59:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-03 15:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-03 16:03:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"59b5d9db6ae8af3abc31956f62770a1b6d6ea4fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-03 14:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9cd83fb33c8fb0b7555cd3de758dcbc8a7912eb","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 14:40:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3302a80d65d87a93ff86925f180373edfd2783dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 14:59:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f9277a9838ac4399807b4b1c38bd67fe : FAILURE in 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6ba1366d6db7543600a9b7f6dbe0fb2607080f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 15:57:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ee074f334d1404691fc4e63585ef9b9 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d0e158c7294b48b96a2dab2c7ebf48 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/629f687b03e045faa2324638f2de1fd7 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd32edecc90e4a5099914eeb2e50e77d : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/617ad81b80b1420a899676e08fd6b50e : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3295f9f00584402297af0315aada755b : SUCCESS in 5m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc629553e63c4f98a20d3c6454bbc1e4 : SUCCESS in 5m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/77c884b49d2448e9ad9d9b2ab146ffbb : SUCCESS in 9m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9fe8a9ac3f4846a6ba18ea25a57c53e8 : SUCCESS in 12m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/93d1970f459b4ca582ec27a175c0a5c8 : SUCCESS in 6m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6c84520347b24325a11f03466419d574 : SUCCESS in 10m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ba3c71023448432d8c58d5225a9451aa : SUCCESS in 7m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e4db6b2867394386889be3cc04510ef7 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13162d94af014fe790fc723f905c1ce0 : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a76e17de21ac4a13803cc7dcc622556a : SUCCESS in 10m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/654a079502af4b9280f24303311d2309 : SUCCESS in 17m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0fc651fb0c794d29a8295eed9d18a7c1 : SUCCESS in 11m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/533382c6a3b345e09ad9440bbea4d337 : SUCCESS in 6m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b0517f3fa4fd49bf8bf7c3b05324a86d : SUCCESS in 28m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0344f20f7ac74c3b8620098df638be1c : SUCCESS in 14m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6ed72895a8ec44eebf4f215c85c107f0 : SUCCESS in 5m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c8d8b34e71848a9b7b0e78b9b765dc0 : SUCCESS in 18m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/33e26b2f24b14e1caf15977b981d0cef : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1a12b275a5b4d249bc6d4ceb89e8a82 : SUCCESS in 8m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/543784f753064957a6d88b8c6cfdbd0d : SUCCESS in 15m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7075a79929ef49419a046cb83f844c66 : SUCCESS in 42m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/865bf3f29ad44f378a412a0a1e1ec3d4 : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e3f11943cb8b685bfdd8506b3374ce82b18a5d1a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-03 16:03:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9d2314f73da8f38de5c76eee9efbf73648d1a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 16:04:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"321bbc4934203c89c8319e4c6de8521d8926893f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:23:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09aa9af7612a41d582d0abfb816cf6b2 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6781a884f0f411ba38d541d78e9809d : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/689bba7c38fc4bd69416d96fa26811f6 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3cadef4b8894fc29ef7b55c460f5f0f : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8eabb6c2af049ab878db3e81bdb1de9 : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c215c9ebce1b456abff34725f098be7e : SUCCESS in 7m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/66a2636241124f92a086552ee61d3856 : SUCCESS in 16m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21b97bfac5d54f298b6605a0e2012120 : SUCCESS in 16m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e2852ddd1ff147f4a8c4d4bc81d96290 : SUCCESS in 9m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ca783a115ffb40f9a2ddb995a1e4f7d9 : SUCCESS in 9m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7a672cef91cb4a629188f398f5a23718 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/89a2a682a87f4fbd8bc7da0362b4ba7e : SUCCESS in 11m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/662fe51b08114735ad11ec67626a840c : SUCCESS in 12m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b236690e8366470a8232af50ff779f77 : SUCCESS in 18m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ef62243fa07c403ba5419c0a27b6d012 : SUCCESS in 14m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a93f8a2216e240fba6713f4984bc4b91 : SUCCESS in 8m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c7b1b350e7da43e882e0973b6ea266c9 : SUCCESS in 23m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/501640ad0d0a43e6823a3734324a83d1 : SUCCESS in 15m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b296063eb5d04ef3bdab5408ce0814d7 : SUCCESS in 5m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5a5b961c3534bceb4f6c7ba99f04261 : FAILURE in 19m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7972a07329a34b55b94e394d0a22c94d : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fcc18687e6cd46ffb92429ffc3e7e9ef : SUCCESS in 8m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4e6742ce0fb943b1abeeb31e5ae2776f : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9288dc617d3447a0a1ef58c256d1c67a : SUCCESS in 50m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea6397f702594128b203d0bb9cdd5f9a : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cbae843ed9484e9f738c9941b0d5f1a56b3758d7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-03 17:25:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b462a0af28b6c332a1f65c83d98070b0a72f5c30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:44:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb4fa8bc2a1a49318aa6d09601d0239b : FAILURE in 17m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b7ec44640fcadea6bc535ea5000043b51c1368ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 18:42:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba326729ad947dda11c073fb94945a4 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72033a63d36f4095bad1a9d8ec59eea1 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f693cf67211c4de3870c66fd754234fd : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2eaafe668e447d09e345b875fec7ef1 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73bd6bee413c467197172ae5b0fa8dad : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c790926f0e1d4584a259d733a245ec01 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d31c7f7be8f440aaab8308cf9f6505f8 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f7f604d20434d84accdfd6b9e200b6e : SUCCESS in 10m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/705afae9bcb749d88d38c8d14638fe77 : SUCCESS in 12m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/71e6aa948ee1417b8023d7480e7fcf22 : SUCCESS in 6m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/af741e2336bf4f91863a5dd951b69cd2 : SUCCESS in 10m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/359ad47cf23a43feb1ad266752fea46c : SUCCESS in 7m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/142ed0422935448d9e2459916accb745 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39e0350c45f04369b6dc64d5add79cae : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c945224a1d3495f9d08a0796cd9ce96 : SUCCESS in 8m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9e58b22537064691b675c9a88b3eccf7 : SUCCESS in 14m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/87690d9bbda84e918f02b86003f96b62 : SUCCESS in 13m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8b42bafe00d7439aa9ddd5090ea7c9d4 : SUCCESS in 6m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/03cfcb5aa33949ed9bdad1e7e171c426 : SUCCESS in 21m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dea7beaeb09f48b0b90e190b4ce3f8ef : SUCCESS in 14m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20221ee818964b589ab97f8e2f309b2f : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b13d91421b8745678e52a5806deba2ed : SUCCESS in 23m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1653edecf22f4fd1ac55b80b685a6641 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbb052860e6440279bc5a4ab28791d76 : SUCCESS in 8m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7931bf2d6a674cceb062f71baaa8ef49 : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/135a5d5c4a6840329861ee919487884e : SUCCESS in 52m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f9428b3a43f4a5e8ffe7a0d90b034b1 : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b2feb3ca58c7117ea694c7266673d4b54227a378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 18:42:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"507de24a80f8d5cea4b07e748447d69e0b9824a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 21:04:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa4a88b1c4b4b71927a2bbe53a91734 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c132d3699f49fc8794c96851c8f2a7 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63df4ddcd9a844c29faa812c1a8981b8 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d3d79c6e6274772bbb03277cd8a8c99 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03fd732e56ad4bd2a92ef5c6b8568a2b : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e0a1f250f2e34855bc508cf31cc8cda6 : SUCCESS in 5m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf9610f77caa44ceaf8c86d0443c3209 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83b9733a78e14243b11e5f9a50e0ce4f : SUCCESS in 11m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d169839bf3e946c28df24f1170df6b3d : SUCCESS in 8m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/486b081b6b82499fbd763170ac1cdff9 : SUCCESS in 6m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d93045a8d524469fb4ef3378c5d8f0f0 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/364f5450f693451e86943e13b645ab47 : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f91a532150f945318ca021ac937ec4e3 : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5430aac45dc24455b777d007da08cc89 : SUCCESS in 18m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/367dc2e335b642d6b57b6387c090dafd : SUCCESS in 10m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/563195d4613340aaa7ac8ca1943dbc6e : SUCCESS in 5m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3a1635118e04413ba0ee6de7fa2f45fe : SUCCESS in 22m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/19a34909aab34aca9d2f50d9cfbb3e50 : SUCCESS in 13m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c80092c7242d48c684ff2ec15562330d : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77c0a2ec3c684a3fa2129cac7c0deee8 : SUCCESS in 22m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9d225a91e52a44f28cb9524ff9798d9e : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d9a331cdfdfb46a8aecc5fbe9a5ec553 : SUCCESS in 7m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/db5d59e5e9d14b8fb77fb7876ee0423b : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5e89e76b089400490f0fd769eda5b9a : SUCCESS in 51m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb0933bbcd2f4e989b7f74c13ad02924 : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a68962b917f580694d5a2de99e412a09f4446cbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 21:04:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"71482458af78198f5fa03d55db14ab5000e41769","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 21:06:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39d3c1efc89d4735a4fcb25972bb78a4 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3dd339f81db217529bd650849325e0c9011db68d","revisions":{"3dd339f81db217529bd650849325e0c9011db68d":{"kind":"REWORK","_number":1,"created":"2021-09-03 14:38:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/807342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/807342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/807342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/807342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/807342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/807342/1"}}},"commit":{"parents":[{"commit":"6d19f0f7de2869f014ff3f23b0e5d1000936680d","subject":"Merge \"bump dogpile.cache to 1.1.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d19f0f7de2869f014ff3f23b0e5d1000936680d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-03 14:38:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-03 14:38:59.000000000","tz":0},"subject":"update constraint for python-mistralclient to new release 4.3.0","message":"update constraint for python-mistralclient to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I08f04104c426134ae678c2aaca89f1109cebf130\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2c86bce636a2aafc00934039d2a19a2365865c23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dd339f81db217529bd650849325e0c9011db68d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dd339f81db217529bd650849325e0c9011db68d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
