)]}'
{"id":"openstack%2Frequirements~807343","triplet_id":"openstack%2Frequirements~master~Ib7c11e16cbc0e956b823656bbab6de52c2a07ae6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib7c11e16cbc0e956b823656bbab6de52c2a07ae6","subject":"update constraint for python-cloudkittyclient to new release 4.3.0","status":"MERGED","created":"2021-09-03 14:41:13.000000000","updated":"2021-09-03 18:37:10.000000000","submitted":"2021-09-03 18:35:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807343-new-release","meta_rev_id":"777cb55024affada056e9ebd941a617524094486","_number":807343,"virtual_id_number":807343,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-03 18:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-03 16:16:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-03 14:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-03 16:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 14:41:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-03 14:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-03 16:09:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-03 16:16:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ef0df1461f325e71180b8e0783e07d52a0053021","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-03 14:41:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0dae36dde20a1863d113316333293670b64e540","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-03 14:41:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"556970e9721147d9cddb39801600ae50ffbdc0e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 14:59:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/407ba0f0f0d04f2294070e526c1384b3 : FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":1},{"id":"390a6717f2efc157692c130d3b315c724f75825b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 16:09:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602a14bd08f440d59230483859418d3d : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb55d3d19ce74f788433394a55ff6699 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6aa42944ac348bcace7e5967bc866a5 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d17f98974444a9a86e01f51728aed58 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec323e13b97d49088b148ad92f828c27 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e62420c97e2645609f2a82ff7880e4a0 : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/063ec2541aae46d7a57bb6cb009e830b : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/000a36fbd0f9423a906cb1c00c9e5a10 : SUCCESS in 11m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bfea88d45de14ae197ee16ea42020ffd : SUCCESS in 9m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/643e958cbb1d4887beb7cbf27f1e3886 : SUCCESS in 8m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d8f3febb20314652846484692b85e482 : SUCCESS in 9m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cc6298aefb9d4a0da58dfdb893dfd142 : SUCCESS in 6m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/62df88dd8a1044ffb838aed2bf5eaacd : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8eb61bea5a248cfa2de04b073787ea6 : SUCCESS in 11m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/80738678068040248f5cce7734ed5169 : SUCCESS in 8m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/abc75d3142a046f3b7747c10f93915d0 : SUCCESS in 16m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/40dd4ca3f4864f89afc5e2aab6046196 : SUCCESS in 10m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7b169c9f6b9d4761a3e75791cf509d30 : SUCCESS in 7m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/468c115f0cee43dd97720eef1187ec30 : SUCCESS in 22m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6434975401e54e03954a3f41a07520f9 : SUCCESS in 15m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e426b074bc474b75b68c8d3660efba0c : SUCCESS in 5m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1901eeffdb6c4e49b6bb8904a0ddcda1 : SUCCESS in 17m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/996cb6281373496dbc6e20d25ed6890e : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d968668eae614101b0a27ace55e1d837 : SUCCESS in 8m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7bf9ca4e30e149789fbab3a0f7faf1c2 : SUCCESS in 14m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe714c41939c4d3da1832f8ef3ba2812 : SUCCESS in 45m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90cdf23554344735b4baa2c5eb47f704 : SUCCESS in 1h 22m 47s","accounts_in_message":[],"_revision_number":1},{"id":"08636562a2979e31c3deb41fd64b3cc8cc883a0c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-03 16:16:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"580226bf6b2cdc808e37598f01046ae09480135d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 16:17:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfac9f76e1b9b7a71a1a410942ed1b5e412c379d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 18:35:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af1e0d2141f4c9ba5bddc4bb4efcde0 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2669d76450a415d8d67764bd8e5bbc9 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc45fc79a37f48b3864787252a96e8b8 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/922ac538f31948b1874a26efae8c1038 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90ed46e61b4c4133aa6d3617135735ef : SUCCESS in 5m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e959c7ce64784024bb62473a62f48cdb : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/071526628b9e4c6fb6bbce54f37e8021 : SUCCESS in 9m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4192e1702d3a42feb7f6bf0043f39372 : SUCCESS in 8m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9188d16445764297ad3593841863c470 : SUCCESS in 9m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/03bbc3b553dd4785bb89411409715544 : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d6223aa5631042f48810406e5072c6ab : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc5d340ac62d4710858d694719525de0 : SUCCESS in 13m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/413b6ccac8ad473ead8afd1a0e265bd0 : SUCCESS in 9m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b787c17df74c47c2a1ce655086a0cac0 : SUCCESS in 15m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1af36184922b4971bc86acd19bdf3d7c : SUCCESS in 9m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/86e489d5053b445cb2c16b788768115f : SUCCESS in 7m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/85e3fb8468004f0ba3bcdb96258d6f43 : SUCCESS in 34m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/127f5dd4f7d845fd9f5a91cb4817af81 : SUCCESS in 16m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56927b8a3eb944cfb27accd0c97d4ac0 : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/262674740f4647ab982c8bb120274ca0 : SUCCESS in 19m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/161cfd9b39d74901a517a0caeabc075f : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a64dcb657fa443fc9354fa28cbde5292 : SUCCESS in 10m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4e7e9426b2ef4e0794fc152334aa7377 : SUCCESS in 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fc5da8c17d54fe89a2851b452376615 : SUCCESS in 48m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c04e35585c4640cea49a3214440a32be : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":1},{"id":"416784240305e0f58104c6e6c2d99428d192b44b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 18:35:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"777cb55024affada056e9ebd941a617524094486","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 18:37:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f0de54af0024005af6d1e4546b33212 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"519a9e4c7bde92c509315254a3e8efb9de5db60a","revisions":{"519a9e4c7bde92c509315254a3e8efb9de5db60a":{"kind":"REWORK","_number":1,"created":"2021-09-03 14:41:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/807343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/43/807343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/807343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/807343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/43/807343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/43/807343/1"}}},"commit":{"parents":[{"commit":"6d19f0f7de2869f014ff3f23b0e5d1000936680d","subject":"Merge \"bump dogpile.cache to 1.1.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d19f0f7de2869f014ff3f23b0e5d1000936680d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-03 14:41:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-03 14:41:12.000000000","tz":0},"subject":"update constraint for python-cloudkittyclient to new release 4.3.0","message":"update constraint for python-cloudkittyclient to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ie80ba4cff4868485bee796bb81586a2702953c60\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ib7c11e16cbc0e956b823656bbab6de52c2a07ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/519a9e4c7bde92c509315254a3e8efb9de5db60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/519a9e4c7bde92c509315254a3e8efb9de5db60a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
