)]}'
{"id":"openstack%2Frequirements~807419","triplet_id":"openstack%2Frequirements~master~I52d47b79b1cb7b30371cce9721b2b148929e44a3","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I52d47b79b1cb7b30371cce9721b2b148929e44a3","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-04 06:27:47.000000000","updated":"2021-09-04 14:44:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"03ce56defcc6b610df569c44a0953660cd063021","_number":807419,"virtual_id_number":807419,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-04 08:22:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-04 06:41:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-04 08:22:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d058f451add9e33986e60c65ade3b7a96dfd4e09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-04 06:27:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd9b1674818ae2c52133a8fdb8da1d9691efd81d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 06:41:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7fde558dfdcd4346b91be4887646b088 : FAILURE in 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"37c4a253d11b5b663e5adce337957decf6b068e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 08:22:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ece1379b6984ff6a39d880b79881deb : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7200703e31184e66ba3c8c44e69e35cf : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c196b2f8d96347e38ff9e545d55d63a1 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64556d8e80344e039527dab44dc823b4 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d03f118068604b22b3ce8339374407d3 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a3b72f1d9ed4881891dd353fa83487c : FAILURE in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4188be1e3dcf466dbfca91f2018a1fb1 : SUCCESS in 5m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b20f2c4e400846a2bbd208deb446a2ad : FAILURE in 7m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5af319ac623e432f969e598c8b18302e : FAILURE in 6m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/24b3810444ba4bedb85dccb0a1b29d66 : SUCCESS in 6m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c47f001ad02043de829587288a511e51 : SUCCESS in 10m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c7e2f3d7f74e4d5c981c831b6fbf24a8 : SUCCESS in 7m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fa75355e90f0455ea286615e8e2bf3cd : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff409a4ad3074322b19316861ec244af : FAILURE in 10m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/53a4e40079fe4961abfe87dd4cb94b4a : FAILURE in 10m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cb978b7d1e8147269a35ad96e146f98a : SUCCESS in 18m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0b75f6b2a9574701a49acf6d702a6404 : SUCCESS in 14m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/970ab11ee2e74e57b93efae2caa30caa : SUCCESS in 6m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/901aa73637c54f1bb5f9bd61667128d4 : SUCCESS in 20m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/23cbcfd6fd4d43d9949a6af274cf5be8 : FAILURE in 12m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8b16bba16d974043900e7e103d29f750 : SUCCESS in 6m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22fe6a5bf2114361902ee78ce2f299e5 : FAILURE in 17m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8d8baba039464cb18fb16fc1a636a8fe : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d9d24e95d56c46f880e6cb79b87e3b75 : SUCCESS in 7m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a48ef3db86e34243b3d39a5985a56fae : FAILURE in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd0e685c38de490bb968c68d7e5731aa : SUCCESS in 46m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a180a9bcdbc245eaa85a40040fa4d309 : SUCCESS in 1h 19m 20s","accounts_in_message":[],"_revision_number":1},{"id":"03ce56defcc6b610df569c44a0953660cd063021","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-04 14:44:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9cf356e1f0d52b06910d8e588925f2cc760a5cc","revisions":{"c9cf356e1f0d52b06910d8e588925f2cc760a5cc":{"kind":"REWORK","_number":1,"created":"2021-09-04 06:27:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/19/807419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/807419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/807419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/807419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/807419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/807419/1"}}},"commit":{"parents":[{"commit":"f405a48c0cee2bd139faa7cf994c5a828940a5cf","subject":"Merge \"update constraint for python-mistralclient to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f405a48c0cee2bd139faa7cf994c5a828940a5cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-04 06:27:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-04 06:27:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I52d47b79b1cb7b30371cce9721b2b148929e44a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9cf356e1f0d52b06910d8e588925f2cc760a5cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9cf356e1f0d52b06910d8e588925f2cc760a5cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
