)]}'
{"id":"openstack%2Frequirements~807466","triplet_id":"openstack%2Frequirements~master~Iea0d6e0389a488d6a8a83a6810e06e8f22664e01","project":"openstack/requirements","branch":"master","topic":"glareclient","hashtags":[],"change_id":"Iea0d6e0389a488d6a8a83a6810e06e8f22664e01","subject":"Remove glare depenendencies","status":"MERGED","created":"2021-09-04 22:39:01.000000000","updated":"2021-09-05 02:33:47.000000000","submitted":"2021-09-05 02:32:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807466-glareclient","meta_rev_id":"4efceaaa62a7ecd7aedcb2d4f5a2e13a5dd6675b","_number":807466,"virtual_id_number":807466,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-05 02:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-05 01:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-05 01:05:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-04 22:55:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-05 00:12:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-05 01:05:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ae504466a737ab1fc70c16a8d0064095db8b3d86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-04 22:39:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"079adae300968992864b88a7142d9a2dc2c1144f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-04 22:46:35.000000000","message":"Topic glare removed","accounts_in_message":[],"_revision_number":1},{"id":"644934bd887741efe0ecb82d0bf4aa9bd321e0dc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-04 22:46:40.000000000","message":"Topic set to glareclient","accounts_in_message":[],"_revision_number":1},{"id":"a3551763e21da51cd695de21f6ad918e97c4ba04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 22:55:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/27079af00a90486ba2271861b90ac725 : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7c8c3263d8d15f8df5f5579a3c037c533df6b518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 00:12:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e086763b606948269a5d0b313a8ad785 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0308eb95f67b4257974dd06d7d24e30a : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9a8379b03844d69b053fc779e8180bb : SUCCESS in 9m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30479f1fd6a64fcb9755b7b3d29a317c : SUCCESS in 8m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45ccf625833a49ce93f33c0541c06a23 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/81eedb4d867e4ac099d84e6fa409a3ce : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e1fd587e29f405cad4c4ef5e7f2205d : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/80d367460a4d434fa8b8259d9564a6a4 : SUCCESS in 12m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73f2e3068f114bd09c6e84fee44f44a8 : SUCCESS in 10m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b684c6e410dd487e8117af1701dc6d71 : SUCCESS in 6m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a021d57b6c4e4dd18cc5c176ca7e9dfe : SUCCESS in 9m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f0397660ab049ef9c8c0a24ac548850 : SUCCESS in 7m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/89fea7d920db4c6bbd247e881a6e4a9e : SUCCESS in 8m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/741ae5761fed4cdb884074679154362e : SUCCESS in 8m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1b761bff6de74d75be5b3cff32fb0d3a : SUCCESS in 8m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/77a23462c8254d70980fc598e948533f : SUCCESS in 14m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/265dd6fa404f4aec8d9ad9a2f80d4ee1 : SUCCESS in 9m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5d387d2d437d472b9d5846f1f773371b : SUCCESS in 5m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1970770afbf542a59b440c3aeb3590ff : SUCCESS in 20m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d2e5109b55ec4ef08dfd3adfef4607eb : SUCCESS in 10m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2462fe122a404ea99957f7f61c667185 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3747a8338f5e41bb97fff41a418841bc : SUCCESS in 17m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fadb1a466fa347f2b64e12dfc7a6eaba : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/92c06a8ff02a44dcb2de68a41b69b194 : SUCCESS in 7m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a211caf49d7e4c9d9c2f69aa8ae46f0e : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16c42f6920d54a428e5acef81344fa1b : SUCCESS in 40m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34101f07fc5d4dfea30366ebef190d0d : SUCCESS in 1h 31m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5bbcca58f165498967466e5fbe2907529d0512b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-05 01:05:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nhuh, I remember when this was added...","accounts_in_message":[],"_revision_number":1},{"id":"70aea4ac876e4652419275c3a743eb8cf9954a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 01:05:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10f5bc6310980e55654362f5772f1e449195fa05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 02:32:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3eb282f9d64722ad68e0e1f1cd97f0 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28df474cb3f4fe196bcddee196b1c96 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abc91fd71d3949d0a5b0b44ad9d73d63 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16a3c03143a84d66a461bf66d05a0e7b : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5177409d9706404f94f7db56cead6955 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bac6f758a6d04df691aee2d8e7aaa789 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/64cc8ea7eac044eeb7e7bc48da27a74a : SUCCESS in 9m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a25054ab8374d528bcb755f71ce9ff2 : SUCCESS in 9m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ef770c471adf4f6faabcddd1aebfb5c7 : SUCCESS in 9m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/70e201e311a74c53b98836a805f9ae38 : SUCCESS in 7m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/294904782a9b487f893adaf24f40f608 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7f2961b71c847698a77f09a92d7279b : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d85b09a713214f908db1fc7d49347469 : SUCCESS in 7m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/59963154cee84dfa831c3dff08c6cf77 : SUCCESS in 14m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d2b7254e8e2a436a9eae0cf1c48841cb : SUCCESS in 10m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/398349e5c23d4ff9a65eb21f53f57525 : SUCCESS in 5m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d9003ebad8aa41ebb7b76315ae658b31 : SUCCESS in 24m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aee2fa516c7043a3b2d338a220acaf1e : SUCCESS in 11m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4361d9b8b7344f5a9dd3452983ae351 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/324647abdb98486086b366999b0a5c2f : SUCCESS in 20m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/95c37cf82c3844bd92c0ad8ec91bc0e7 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/193774d0f6c149faa2909353dff31ebd : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dbd75431cec04ec084bdbbc1966b25e4 : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef5205f1fa184f06b530a0a280d4dd18 : SUCCESS in 45m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86bc20a2954241cc82cbdca165b3f6d7 : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"bc59f78ed4a33d5f4e5123e4d7a539cd0ebcf026","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 02:32:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4efceaaa62a7ecd7aedcb2d4f5a2e13a5dd6675b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 02:33:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f274b3631d3423b8356b020e5ac8d2e : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6dd9373b238b9af85a8bd3075628875cd4a6f112","revisions":{"6dd9373b238b9af85a8bd3075628875cd4a6f112":{"kind":"REWORK","_number":1,"created":"2021-09-04 22:39:01.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/66/807466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/807466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/807466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/807466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/807466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/807466/1"}}},"commit":{"parents":[{"commit":"f405a48c0cee2bd139faa7cf994c5a828940a5cf","subject":"Merge \"update constraint for python-mistralclient to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f405a48c0cee2bd139faa7cf994c5a828940a5cf"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2021-09-04 22:36:34.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2021-09-04 22:38:52.000000000","tz":-300},"subject":"Remove glare depenendencies","message":"Remove glare depenendencies\n\nThe Glare project was removed from governance and has not had any\nactivity for three years. This removes python-glareclient from our\nglobal requirements so we don\u0027t test compatibility with it any more.\n\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: Iea0d6e0389a488d6a8a83a6810e06e8f22664e01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dd9373b238b9af85a8bd3075628875cd4a6f112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dd9373b238b9af85a8bd3075628875cd4a6f112"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
