)]}'
{"id":"openstack%2Frequirements~807544","triplet_id":"openstack%2Frequirements~master~I7111787d2a45cc10a5d5d4dffd4e338b1e1d5232","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7111787d2a45cc10a5d5d4dffd4e338b1e1d5232","subject":"update constraint for futurist to new release 2.4.0","status":"MERGED","created":"2021-09-06 08:31:37.000000000","updated":"2021-09-06 18:13:32.000000000","submitted":"2021-09-06 18:12:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807544-new-release","meta_rev_id":"7ba726e5491d31c0303d110aed13cf5ca6e4dc75","_number":807544,"virtual_id_number":807544,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-06 18:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-06 16:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-06 08:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-06 16:25:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 08:36:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-06 08:51:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-06 09:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-06 14:51:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-09-06 16:25:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3b9759effd6ab488dd1209daeba359d89499e345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-06 08:31:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad6f05fc0b1c4cf27a1163ff2ee28b9e2e9f3bbc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-06 08:36:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"286f839668f70d1b3a500d1e9777bb825265b8ca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-06 08:36:52.000000000","message":"Patch Set 1:\n\nindependent project","accounts_in_message":[],"_revision_number":1},{"id":"472030b0b868ae0aa6e1e6be1283aba42b930e86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 08:51:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4d2ec4c017bd4f86b7e4c03990a2d450 : FAILURE in 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"22ce35606490d1b894c0192e799a2b3a13e5431a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 09:25:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5779da751c8a4f88a63a0a765ebaf629 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7d90df460ca41979ec71b0891f2b10b : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5809847b0c6a4c5db840736c9b94a988 : SUCCESS in 7m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d81d3089756a467b8fb050a3b62d06b0 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74d6be91e03e43979cd71ee2fb2e875a : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aad6df91d3184cbbaa34abf718881fd3 : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f73298105deb47598877b9e9e16f1f73 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2fe5638839c647a880039f5251bd3b40 : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/faff44c27c834e1081f225fa98a1d09f : SUCCESS in 12m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3583867ea99b4aef9d3ba56c86dc079f : SUCCESS in 7m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f2faac6df20d47db9c9bcfb4df7231e6 : SUCCESS in 10m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3389c770087e438fb8c9bb00b8a49f46 : SUCCESS in 6m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5aeff883cf4949b395288d2bb99a3ea7 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e268b701426d4733b6c4497d6c95fa89 : SUCCESS in 11m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dcba74a3d0584073b694f1d03aedc040 : SUCCESS in 8m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/847cf709fe8949ab842d5cff2ddd9238 : SUCCESS in 15m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2c49fa6e70a9496b80d45691c451eef5 : SUCCESS in 13m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/75de26e4a76a4af4b2c9b7b2a0613522 : SUCCESS in 6m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9d1b3f77b8de4eccae58b32097568e08 : SUCCESS in 23m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8140e33dbc7c4fbdba33bc6a8172fe79 : SUCCESS in 12m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0af322d5159d4aeaafcfaddb348554bd : SUCCESS in 5m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0711b86774054403bc923bfe44ebdba1 : SUCCESS in 20m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/39e8fb0106eb44a5b8b2cb1ce88e0e62 : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71eee84cbb0344958b544ae75dd76ac1 : SUCCESS in 8m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e33045db232e42c9a62d5e70c13956b8 : SUCCESS in 15m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f3c41397bd54c73aa7bc27fa73885cc : SUCCESS in 46m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d104620c729c4505bfab349aabdc4f28 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"815107589dc8ebb7736be0e21ec2c62f79886782","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-06 14:51:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25e6bff7a046e438272e4a9fab115091a7896fbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 15:10:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3b86540fed2f45139e0fa87a4bf401f8 : FAILURE in 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cce04905f24eab549dcfcdee6d3e8006e7c81b81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 16:23:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0d573f14e6848dbafce77ddc9fce3cc : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc5eb7ff8444d9e959a0e755b12c897 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6ef3cefe7334e6c8612d1116cbf3b14 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2af56097250f4d0eb67be8cee69a08f5 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b42cfb348ed74f498d61fd05be5b72a6 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/172699bb31cc47a2b5be0ef84ba5f9f9 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d04b220b37f4cb0970f335cd2477223 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91bda592862d4813be4080e869f35619 : SUCCESS in 10m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/404562379c28417cad1c55fcdf67e52f : SUCCESS in 10m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a60dbae8077f43219c79848042e5aa41 : SUCCESS in 6m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bbb0ebc4b00c455b9806b6f39044f63d : SUCCESS in 10m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d5d96f87bdc1404ba26e276c71742bd7 : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2878a7bdd8bb45238b8e1d82460b4ce2 : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e855647f85b4e1aa759f2c7ac615a22 : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b329998713b349bbaa1760ea00ddddeb : SUCCESS in 8m 53s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/27bec68ee8b546a38d5c02ca58d75ec3 : SUCCESS in 20m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/912ccfde720944d3a9ef295e703179d2 : SUCCESS in 10m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4e54096041a2415e85528c30bd0ee860 : SUCCESS in 6m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d336e3c958b1414abfd642482053b397 : SUCCESS in 23m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e8b0c69eeef14347a98ceee256daebce : SUCCESS in 12m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/60a87a24c5bc461dbb1df40608bb4ae0 : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c9235978633d44c2ba7f20cec801b0ea : SUCCESS in 17m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a920faaaeaef477391a200ee23ac920e : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59eb9e90547f4e3a8775dbdc1fea69ce : SUCCESS in 8m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/72d799c04f9745c6a1691dfebe4f6565 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/984f10e8b6c142b9a6d060d93e6eaa45 : SUCCESS in 49m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5b5bb1dca1740358f7718265cd7fe29 : SUCCESS in 1h 24m 19s","accounts_in_message":[],"_revision_number":1},{"id":"bfbf01b815d0c6bb9e31a0701401be1d1099b15e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-06 16:25:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c78b636b20c9178c4523d831cd35a0452196836b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 16:26:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"084297f3f9df2e658bae75b46a7e57495eca6294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 18:12:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2f7426bbc2497cb942c9d0b11fced6 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5288567bc57b4f92bd4f7961f718fcde : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8172325f379f4f65bbf7669ffdc6c0ba : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70c4091b731845edbd3482222813c30d : SUCCESS in 6m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc867d29f4e342d5a1fe2ea084b27c53 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d02a68e8a441407aa32a35afd50cc4cd : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/81df9fdc3a4e49bf8b2ac810acd4f51e : SUCCESS in 9m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/861f16e6011543ba84547460e69187fe : SUCCESS in 13m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f44ccf7eb8c84bd7a1c1bcb52ef33ceb : SUCCESS in 10m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a15260e2437f4dc1b0925a2a2ad2bc6a : SUCCESS in 6m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c1303e91af9a4f0b8232bf9c95b93673 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f66f89cb6af74c47a3ce76866033e288 : SUCCESS in 9m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/deb547a8b6cc4c63a0ddcf8c520294fd : SUCCESS in 9m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/48dca9eb0945400b8b195d103ba975cb : SUCCESS in 17m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/03d4e85ed8564138b28ab9c19054e14f : SUCCESS in 10m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/01bce32180744ef3af214adfae53ce77 : SUCCESS in 6m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/22ccfe716fa243f4b2f50c1531d001f6 : SUCCESS in 24m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0ac9cab5cfa5401d96e66b23030a1b74 : SUCCESS in 14m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee5abf60c1334b9f99f392be9471f0e3 : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/42b3bc294059429c8760e90e067eac3d : SUCCESS in 20m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/17181b70987641a0812e6605f9fe8050 : SUCCESS in 8m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1d071232d644410855ed662bf9861c3 : SUCCESS in 9m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/02978e5570af4ad3b6b0618d54f290c7 : SUCCESS in 14m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d7f0da43eaf4b71b4688f40bf1da212 : SUCCESS in 42m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/686fd6ee6666438ea39de812bd2bf562 : SUCCESS in 1h 25m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7fcaae5c77d9d80b15d26f7f05a04ba2b89af06f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 18:12:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ba726e5491d31c0303d110aed13cf5ca6e4dc75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 18:13:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89388531bc114a00a5cd0d2c91f9d693 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"beb78825ccd1daea0b839033091f0a991e6966c7","revisions":{"beb78825ccd1daea0b839033091f0a991e6966c7":{"kind":"REWORK","_number":1,"created":"2021-09-06 08:31:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/807544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/807544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/807544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/807544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/807544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/807544/1"}}},"commit":{"parents":[{"commit":"6dd9373b238b9af85a8bd3075628875cd4a6f112","subject":"Remove glare depenendencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dd9373b238b9af85a8bd3075628875cd4a6f112"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-06 08:31:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-06 08:31:35.000000000","tz":0},"subject":"update constraint for futurist to new release 2.4.0","message":"update constraint for futurist to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I8582f7e7dbfe22189b8221392f6ddfe415d816b0\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7111787d2a45cc10a5d5d4dffd4e338b1e1d5232\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/beb78825ccd1daea0b839033091f0a991e6966c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/beb78825ccd1daea0b839033091f0a991e6966c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
