)]}'
{"id":"openstack%2Frequirements~807548","triplet_id":"openstack%2Frequirements~master~Ieb0e82e611f2d124c8289309b4b9368e86618771","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ieb0e82e611f2d124c8289309b4b9368e86618771","subject":"update constraint for debtcollector to new release 2.3.0","status":"MERGED","created":"2021-09-06 08:38:31.000000000","updated":"2021-09-07 02:40:42.000000000","submitted":"2021-09-07 02:39:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807548-new-release","meta_rev_id":"b16d24081045996d592ba60348e8e1f572c55216","_number":807548,"virtual_id_number":807548,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-07 02:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-06 14:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-06 08:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-06 14:50:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 08:39:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-06 08:57:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-06 10:05:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-06 14:50:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c3d3b8f04c01778cbd606ae03900810ac0c56fc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-06 08:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e77f65a9235a4f141d200029cfc34127cc711331","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-06 08:39:41.000000000","message":"Patch Set 1: Code-Review+2\n\nIndependent project","accounts_in_message":[],"_revision_number":1},{"id":"34d3f4180422caf4a756dc8222256917cd37bceb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 08:57:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aaf7d492d68c46008c5a02e00f5cb2b6 : FAILURE in 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c57a5f481407a373a028cf0fa2bbee8022674812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:05:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3919450348b94696a47aff39c78d23d1 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20dae47bae56430ba869abbd92658d7a : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f070b6aff800469a822a4d88170f44de : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b13112df16714a2db0b90c7a14da14d3 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62b045fe11344d2a88cb00d46aaf7fc5 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be72e53512ff4c7582cb0da97edea861 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4a2487322f647bebad7926df6b4ba9a : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/042eebb5d0c6417eb9eadc40ff87637f : SUCCESS in 10m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/240ea4775a4b482d8fdea78155304515 : SUCCESS in 9m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/900f2fae096643ba8d4c89700f097edc : SUCCESS in 6m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a10c4d3a73a346f3bb533936824180d4 : SUCCESS in 10m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/142daa29263b47c1a454106428404872 : SUCCESS in 8m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/714e3d93377f4030bcf81704cc27201c : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86f2135999e6488eaa3d03a3c0695fa6 : SUCCESS in 10m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2f808f9648f24b7eb7add91fbc434b7c : SUCCESS in 11m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/16bca1eb407f409388fa9052486e6ca2 : SUCCESS in 16m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cfd85d431a4945a0845790ef0f805984 : SUCCESS in 11m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/08866b99351c474d90c27991823d94e4 : SUCCESS in 8m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/68201db0673e47b88d54cbd25d8082a2 : SUCCESS in 24m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/302ac46b48574940850a77eef75d515e : SUCCESS in 12m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a934c84a3d674d6d8c795aadd7470d3a : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7dd9b232398f47f6af0d51ecded471ac : SUCCESS in 18m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c6a99b8fe1204202b3572fca17470548 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d46a7aaf655f40d9bbd655d071a224de : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/160815a87bfe42fd9d953ab40d28eac6 : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19e1636071354a4c9f80796ac25e8447 : SUCCESS in 48m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52a96114191f44a08b905d5ac22d05b1 : SUCCESS in 1h 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"2f1fc080a568c7fcb193ee47405636fadeb26da3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-06 14:50:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"faa526645bc3135ddac969eca7b87d6a843d3b11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 14:52:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f9b34e56756026ad5961ee3cf0352d2207dcb33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 17:11:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845b7e12087f474a9f9b2e3de1440097 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d7e728b94ea4957ad9ea40cd8d5e74a : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3095847f216b40c88bd7d870e222c091 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4185f922938402e8616a735251eed2a : SUCCESS in 6m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e5eba3494c984a3da8c7ea5087953c06 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a1e58fc48d84635baec834d30837b0b : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1d18151bc70a48969d4cd0fad77c3e9f : SUCCESS in 10m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21bf6d386b364646970cb9d7ccd2aa2a : SUCCESS in 10m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2c5df3d3511d49ef849f7ef427f00f55 : SUCCESS in 13m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/17e92e40a78f48e9a3e8bb3a877802c0 : SUCCESS in 7m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7ac2c4c7ff98409489c63789c5738265 : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c714cfe435e4d9eb1a482279789fb73 : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ad532d7c90ff49b19d813de8ecd696ff : SUCCESS in 8m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f032d1900e6249b5a9886d5c39297bb6 : SUCCESS in 15m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7d9388771d994e30a2cdc0088fa12099 : SUCCESS in 10m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5196f41dcae2480b8aba8211d7e52f5a : TIMED_OUT in 41m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4d32a9c240c2463087b81753c8771dbb : SUCCESS in 22m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f0ab0ff75ecf49dea2dc2808461c5eb0 : SUCCESS in 15m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/17ee8a91894541e8a4e0a6b13ff1a820 : SUCCESS in 5m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6058dad6ab524fdab7c83cc78dccc013 : SUCCESS in 21m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7dcb61affbef4beb96eed1d4c946690c : SUCCESS in 9m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de060e53d15c430e8c75590885c3a8bb : SUCCESS in 8m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/178cffa98dd84fbd9f0ae916511ae16b : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dc95a9c86fc46f3921b2177f164347d : SUCCESS in 49m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc84668348b348b88dc62384ea924b6b : SUCCESS in 1h 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7558f6ff496e2f1be1c6a845ec58394a6efb3b10","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-06 23:21:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1cee5b8a8e6b45696a645a468239fb131b6e2e9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 23:39:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e5b2ee9c35a54d7297dcd192d7c03971 : FAILURE in 16m 56s","accounts_in_message":[],"_revision_number":1},{"id":"38970bd3e0e532f7e441af18a25a6e9a8e7fb3f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 00:52:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/326da2742d27416da71074d4424e5cfe : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176c73fafb3f492586a8c93b5968b46d : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7259756aec6644a98ef0e41f21b64bb5 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2de865241ff245769b6aa4c33074314b : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c652e7ac19349c9b5d49071945343cc : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3971cde661484f10952444107a1e89cd : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1cc71333a7a742d3acda5d15026aaa88 : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e021e3b48704cf19079ed8f35a06ff1 : SUCCESS in 10m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ed6368888b1479d8bbc7589dd534dd0 : SUCCESS in 9m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bf59f580c5c34d6b837a25ff166ea87e : SUCCESS in 5m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9d8783ad28e34e889edcf253d84beaa8 : SUCCESS in 8m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f014ac8bdf63465c8ce7dee3e0032323 : SUCCESS in 5m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/51800aebde6d4e0cb79e221e3e9c251d : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d71f68219c7a4a198845f228a182b2f0 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b71db301f357474c87c53ed3f760dd00 : SUCCESS in 9m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7ae2c053b72f4b7c9abca052b41ec9f2 : SUCCESS in 17m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/711a33766b02443ab139768d0458f5a2 : SUCCESS in 11m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0a046bd653cc41479157f75b413ab461 : SUCCESS in 5m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/247d8b050216451b92025cca039d0a04 : SUCCESS in 20m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1d1271bde5c7437a8835f65ad01097f7 : SUCCESS in 10m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3642cd5ade6e4fd3b08a6b1fbbfb2686 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d37889b28b14d5e90a33c5cd069872d : SUCCESS in 16m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/df1d4ffd30e04654a0efed157f2295f7 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/061a9c7794b54b69bc736fb5c4b60bc5 : SUCCESS in 7m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c286c75bd9a24efdbcac4567f9f557cf : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7646f6ccd8845de8b93149d890ba00c : SUCCESS in 42m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a0cd8e1903c449aa70b6481674f576d : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"939c17dd300fbabac6172f3f2143a632a18fc3be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 00:53:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"216dde0a1e60a3373598143881c90ce60100c64f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 02:39:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47b67d3b26444ed987497d1a55bef50c : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b74dd633f7540aa9c3b2e63eefd5be7 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6827eca58824ca3964f3ee2bc9e9db4 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8af742731d747a1bb9e4c395fbd5fa3 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c85c4c41a7549a0a5f4797cf9f5eedf : SUCCESS in 5m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/acfe0a3a6de140c9a0eafd75f6f9a98a : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/52d531dccfab4bc8b88f9f6325eae27f : SUCCESS in 10m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/666c252b118a4491ad4cc5e76a70c70b : SUCCESS in 14m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/de0c313dc9804acba993768452bb1115 : SUCCESS in 12m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fc85093e14634d70b9686cf816f79913 : SUCCESS in 8m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2d2dae5bdcff4d22ae6bf01234423f76 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62345b232dcf47eea3cc2783ad3ed89b : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3f36cc2de3654d198bea0eba69fd4143 : SUCCESS in 7m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cffa6f3397fd4e28b1cf720274bf9309 : SUCCESS in 15m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cd831421e78f4d3a98657401e7b9d02b : SUCCESS in 10m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/873a95b447c14600bdeee419ad6ce187 : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/52464f2e8f9c43ddbdc62f9346138031 : SUCCESS in 21m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/329ad7ca12bb4f30a6d9e39a3d5e77ae : SUCCESS in 14m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/090d7f2bc302443699fc275e083ebddf : SUCCESS in 5m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22f8d33c19df4dd785d7f8e82105dce7 : SUCCESS in 19m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a21e7f2c989f443c97df593be69a2ac7 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/597bd52edf9d4487a7ef84283087e871 : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b14a29eb374e4be9be5e513fdb630016 : SUCCESS in 15m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d85ea8e3aa6e4d33b5c49c5dbef2a2dc : SUCCESS in 38m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb83198ec4334384be7869398cead4c8 : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7c4c4bc0c13c9e5774b571f350611e8ebe63f7c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 02:39:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b16d24081045996d592ba60348e8e1f572c55216","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 02:40:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b409b034e74b25b5c6c10dc11377da : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7fe858b1b614208ee0394d34c9cf0c72363fc79","revisions":{"e7fe858b1b614208ee0394d34c9cf0c72363fc79":{"kind":"REWORK","_number":1,"created":"2021-09-06 08:38:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/807548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/807548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/807548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/807548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/807548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/807548/1"}}},"commit":{"parents":[{"commit":"6dd9373b238b9af85a8bd3075628875cd4a6f112","subject":"Remove glare depenendencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dd9373b238b9af85a8bd3075628875cd4a6f112"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-06 08:38:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-06 08:38:30.000000000","tz":0},"subject":"update constraint for debtcollector to new release 2.3.0","message":"update constraint for debtcollector to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic67b969426a65c7ada241e98b6a8135d18ff23cd\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ieb0e82e611f2d124c8289309b4b9368e86618771\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7fe858b1b614208ee0394d34c9cf0c72363fc79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7fe858b1b614208ee0394d34c9cf0c72363fc79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
