)]}'
{"id":"openstack%2Frequirements~807571","triplet_id":"openstack%2Frequirements~master~I36331013568a36c963c180db10e30c3b72cf86a8","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I36331013568a36c963c180db10e30c3b72cf86a8","subject":"Adding ansible-core to the blacklist","status":"MERGED","created":"2021-09-06 12:10:06.000000000","updated":"2021-09-08 18:29:40.000000000","submitted":"2021-09-08 18:27:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807571","meta_rev_id":"59269e1db434bc8ebe4e0baf1f2b59a29b4ad2fb","_number":807571,"virtual_id_number":807571,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-08 18:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-07 14:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-08 15:28:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 13:26:22.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-09-06 13:26:22.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-06 13:31:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 06:06:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-09-07 06:06:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-09-07 06:06:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-09-07 06:06:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"}],"messages":[{"id":"a24a8ef4d8d536a9111a20d230bfbfe30f4a1a68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-09-06 12:10:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea4e012dc82f53afc2dd3b6b95c5f2fa41c6f02f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 13:31:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544e63b399a943b3ba69475a51d1128b : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2928a3a9187457887454feaff028a93 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19700fb6a89549638888fc732ff0e962 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d27abd61d46420784f3afcbdcc90435 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a990bf27812f427db08d523d0e91b971 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2444da8cb06b4e12a7298d4dc382f0c6 : SUCCESS in 4m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc8513bcc90640eebb83e705ec3da25f : SUCCESS in 39m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/877fa28cae62453a8b2d75fe3417ede0 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":1},{"id":"abf281a16304444b990652c8c5591bb8b7ce4252","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-06 16:01:28.000000000","message":"Patch Set 1:\n\nI blame redhat for this crazy rename stuff in ansible :|  Should we remove ansible and add ansible-core instead?","accounts_in_message":[],"_revision_number":1},{"id":"645654e728087e13b6e1614577971992baa9b89d","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-09-07 06:06:47.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I blame redhat for this crazy rename stuff in ansible :|  Should we remove ansible and add ansible-core instead?\n\nI\u0027m not sure removing ansible from the list would be a good idea. The change [0] is rather recent and not all projects move at the same pace. That being said, it will probably come up at some point. \n\n[0]https://www.ansible.com/blog/ansible-3.0.0-qa","accounts_in_message":[],"_revision_number":1},{"id":"8427fa6e09301744d42b60916593bd0d7d016353","author":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"date":"2021-09-07 14:14:06.000000000","message":"Patch Set 1:\n\nWhat does this actually mean?  What purpose does this blacklist serve?  Can it be named more meaningfully?\n\nI have no idea what this does.  My only clue is that perhaps it\u0027s a list of projects that are _not_ incorporated into OpenStack requirements?","accounts_in_message":[],"_revision_number":1},{"id":"f3b0561e123c957f74665ac8104ffa72e50b6d04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-07 14:22:46.000000000","message":"Patch Set 1: Code-Review+2\n\nblacklist is for things that should not go into constraints specifically.\n\nhttps://docs.openstack.org/project-team-guide/dependency-management.html#general-review-criteria mentions it (just a single item in the list, but still...)","accounts_in_message":[],"_revision_number":1},{"id":"93268fdc8005306f39c99169a9b4dcdf5412342d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-08 15:28:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c666fe2121cfeb84b035f22d97f4a0236e8b156c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 15:29:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bdc82fb1508e17c41816aae88aed8241c2ef0c59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 18:27:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55cb83bbd63c4c66a58a10fea193cdf3 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/143f397858bb41cb9fb7c2e2e326768c : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4d1d4422d6740b8a716ef977f57e933 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fff4140ee5d34261a621cd55c664d7f8 : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14bb33c1ade444a3a60ec06f03a81389 : SUCCESS in 4m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71d466d14fe64e7a83304c5917d83900 : SUCCESS in 51m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08c5bceeac6e41bcad30ae46d4dadd58 : SUCCESS in 1h 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8e0d48db549c84f313f8bd0a36cbead6c9a8b685","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 18:27:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"59269e1db434bc8ebe4e0baf1f2b59a29b4ad2fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 18:29:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1cd7032c73c4991ada37755bc463f06 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99881be36ec7827a87cde5dd0e06142c8cfe0154","revisions":{"99881be36ec7827a87cde5dd0e06142c8cfe0154":{"kind":"REWORK","_number":1,"created":"2021-09-06 12:10:06.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/71/807571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/807571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/807571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/807571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/807571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/807571/1"}}},"commit":{"parents":[{"commit":"6dd9373b238b9af85a8bd3075628875cd4a6f112","subject":"Remove glare depenendencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dd9373b238b9af85a8bd3075628875cd4a6f112"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-09-06 12:06:44.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-09-06 12:06:44.000000000","tz":120},"subject":"Adding ansible-core to the blacklist","message":"Adding ansible-core to the blacklist\n\nAnsible core is going to take the role of ansible in some projects,\nas the dependencies move to the ansible\u003e2.10.\n\nRight now that\u0027s the case only for VF, but more might come in\nthe future and it\u0027s best to be ready.\n\nThe inclusion of the ansible-core on the blacklist would serve\nexactly the same purpose as the inclusion of original ansible package.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I36331013568a36c963c180db10e30c3b72cf86a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99881be36ec7827a87cde5dd0e06142c8cfe0154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99881be36ec7827a87cde5dd0e06142c8cfe0154"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
