)]}'
{"id":"openstack%2Frequirements~807592","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I36eefb414fe05288c1454d8caf7dc73a7121396f","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I36eefb414fe05288c1454d8caf7dc73a7121396f","subject":"update constraint for castellan to new release 3.6.1","status":"MERGED","created":"2021-09-06 14:51:02.000000000","updated":"2021-09-06 18:12:04.000000000","submitted":"2021-09-06 18:12:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807592-new-release","meta_rev_id":"000322e5730436c7cd973053cb6c4abd56fa5c6d","_number":807592,"virtual_id_number":807592,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-06 18:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-06 16:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-06 14:53:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-06 16:25:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 14:53:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-06 15:17:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-06 16:20:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-06 16:25:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f3cbf302a1e20e0f7acd003cfbf742659d5afacd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-06 14:51:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a043b898f657a99598280ba064372f0f201917c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-06 14:53:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f76653863a225b5650b2380873b8c7822cba462","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 15:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/81f32ee89e4e48669dd7f82ac10cce72 : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":1},{"id":"83dfe3a7c497258537a26fc3c317a83b3d2361f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 16:20:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e02129a3484dc086fbd32937ea007b : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7359e457f80b49ec8526dc75990c610c : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe8eca9c5d3f4a4581d2f144c3871caa : SUCCESS in 7m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05451a6a63df4020a66fdd6e46a45b9c : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ba9d02e80374a61935dfec2b5eae52a : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2f6cb2131c1c406abe61b2c81faf49bf : SUCCESS in 9m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/adb11a29cb7b416cb779dd335ad36192 : SUCCESS in 12m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b41fd657a45f42aa98a861cf52862c29 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/13f8129c40d547d5b219740916cb2171 : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cb6e23d68ca44d608725719737d97958 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01ed85f71d50432e9c9179762d87ba47 : SUCCESS in 11m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/50d0c3f51111407f8b7be0bfa22acf3c : SUCCESS in 8m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0300626e8bbb42be9485b402c4266e56 : SUCCESS in 14m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/738b0ff5420741469f4d21a426b95991 : SUCCESS in 24m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/73268418be424408b0c37e31c1d4fa48 : SUCCESS in 13m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2bc9eea24e614be8930c5401e73b4d5c : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/86611b60e0ac42dd988c1365b48b8faf : SUCCESS in 19m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7e10728d321344f78113883025694afc : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95b5b0b449c44cfb826edddfe5df4983 : SUCCESS in 8m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5e8d493583ca446c8f815aee761b7328 : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64c2d1d1c6c64aa5a06b6e1e71defb60 : SUCCESS in 55m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/368dcb2db22f47258d649d5294c0eee8 : SUCCESS in 1h 25m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3e5519a03bb158fc98683a590df9a58fa34a72f8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-06 16:25:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d533c8bfc55407b7225515a2f5b2abde5586d641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 16:26:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d23c4020f06aac6cc7d789c221bc246515c8a367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 18:12:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee0ee4a60154438e8e8d2e5c8d824664 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e29cc6a10c84456e9b55c60905cec52a : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f75d234feb574384a31e942064e973a0 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5a813be8ff74a82af7de0a5f9ff2f3f : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2943e35224141dfb74de250c1973274 : SUCCESS in 5m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/111fcf9fef5f43b68c02b7f22d23a168 : SUCCESS in 9m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a669c185a1f94ec2b2463ab2371635c1 : SUCCESS in 14m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a82c7d84bab74f93b7390371a03e7beb : SUCCESS in 11m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b269ce016384f64bcefdcbb76a9fbfb : SUCCESS in 6m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/472aa827f67a49d1805fc84519caa4b9 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c72caad26479457cb9f4a592654e5abd : SUCCESS in 10m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3960fd9ab77346b98c823ae4dc48d610 : SUCCESS in 8m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/78dc0600c54d4a0897da4f6a0d52b388 : SUCCESS in 23m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f595bd8d2264477e857c87c1ae9a82ef : SUCCESS in 24m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab53cd11a4074f389eca6709ce5787e2 : SUCCESS in 13m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de01a6bf19c1401d86abd3c09c39e14d : SUCCESS in 6m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b52f941b86c476eb2bd77be4ba34707 : SUCCESS in 18m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4878dcb82da84f048a838a3a4379e646 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7ac5d5744ed4322a5c4f759b6a8ecca : SUCCESS in 8m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3a26c9e3f85c4c20b213693e7674e4b0 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d393e26685534a5688e2d83a70713d02 : SUCCESS in 52m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/981d7ba43e124e64a5e7cb964fb3609d : SUCCESS in 1h 39m 22s","accounts_in_message":[],"_revision_number":1},{"id":"000322e5730436c7cd973053cb6c4abd56fa5c6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 18:12:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36f6b2bf3cd921f36d5bf9bf675cdb1b6f8d40a4","revisions":{"36f6b2bf3cd921f36d5bf9bf675cdb1b6f8d40a4":{"kind":"REWORK","_number":1,"created":"2021-09-06 14:51:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/807592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/807592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/807592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/807592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/807592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/807592/1"}}},"commit":{"parents":[{"commit":"19b608293e6c9f114d55b50e120fabc78fe24276","subject":"update constraint for tripleo-common to new release 13.2.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/19b608293e6c9f114d55b50e120fabc78fe24276"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-06 14:51:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-06 14:51:01.000000000","tz":0},"subject":"update constraint for castellan to new release 3.6.1","message":"update constraint for castellan to new release 3.6.1\n\nmeta: version: 3.6.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I3158e734373754aa8793d0ab2b1bf3166cb60551\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I36eefb414fe05288c1454d8caf7dc73a7121396f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36f6b2bf3cd921f36d5bf9bf675cdb1b6f8d40a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36f6b2bf3cd921f36d5bf9bf675cdb1b6f8d40a4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
