)]}'
{"id":"openstack%2Frequirements~807595","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I6e2570a231bd4d5d542709fa857c7d7da2024a25","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I6e2570a231bd4d5d542709fa857c7d7da2024a25","subject":"update constraint for oslo.db to new release 8.1.1","status":"MERGED","created":"2021-09-06 14:58:27.000000000","updated":"2021-09-07 00:40:52.000000000","submitted":"2021-09-07 00:40:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807595-new-release","meta_rev_id":"67fd6d986f3c6836d3b349666b63d5ada8d0c1e1","_number":807595,"virtual_id_number":807595,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-07 00:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-06 23:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-06 23:21:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 16:36:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-06 23:21:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"16cbfe2809e2f02b17d38ebb47b982b09d754631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-06 14:58:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56cd1ac41b7a12b93bf01c7492c4429dc297a775","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 16:36:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54c453dcccf34105a6ac50d1b778f10b : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9e0597026e646cc9230d54d562159a9 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58c635ef8f504b4e9c81af7e9faccede : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9160d8f9899a4b888bc6c1f3c9d6ae69 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d073a9728a9444feae4fd55093602ed9 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb453821097c4ecc9a04bcdee417a9f3 : SUCCESS in 9m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3c92d21baa944bd9b67c84a6edbe4fda : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40bf73578181413695f2ae2b515504f5 : SUCCESS in 6m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e8af23c6481e4ab2a986e23f6ceb6885 : SUCCESS in 12m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8ce2f639af34a0f9e4e8622e9c963c3 : SUCCESS in 8m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9d7d24e3527e42508262bd5080c49a55 : SUCCESS in 13m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2e8d378612794f23b8f99235196b00c9 : SUCCESS in 23m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/556fe38574d04913bfa19a79589944c5 : SUCCESS in 13m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/359594c03d5040c38f36ae252525af34 : SUCCESS in 14m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37ea970b404347958055cb76c3e41710 : SUCCESS in 18m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/43aaaed411e542cb9b31d0c8bc7f9fca : SUCCESS in 6m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9aa58aebe60b4becabb5813ac1a63d83 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d385ad4dd548413c899a3216ec4720a8 : SUCCESS in 8m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0d3c64099c4f4b88a8920034bcd8343a : SUCCESS in 11m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2eeca377440c42c8aa5721ffb07e3ed1 : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca1c11f6ace44bdba2579b1dfbc6f4a6 : SUCCESS in 1h 00m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bad4b921c5f349a98704c34f994aa6ad : SUCCESS in 1h 09m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e3636eddd584b2baa5b1dc7e9a9447d : SUCCESS in 1h 28m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/b054bca56dd74664b37102071facdf0e : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"503cc73c1872e156d09c741be195ffd92418feed","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-06 23:21:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3dd0870f76990458d114c07840bd963381b98dfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 23:21:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0d472dfb9c1026cb62d370ea93ead7ee11951ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 00:40:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf51d41de1eb4050b815c887c104ad0f : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba1acb57c064f86ba69b915726924c6 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36fcaff9a23a448c8e664874c50ad1b4 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a97ea8d2bdbd45058fea961821586111 : SUCCESS in 6m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f983d561ec94453bc255a50eb765d3b : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e5806056a35c4e2d86caa31789ca5a05 : SUCCESS in 35m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1b8d119d3dbe474b9a17f00b717bf191 : SUCCESS in 17m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/549467bb0749482aa7d9de4c6046f6b3 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/99e1a56f024a4c098402526b6d161427 : SUCCESS in 13m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/358657979eca4d2a98528c54199ad006 : SUCCESS in 8m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/40d035863e674afaa16c1b673a1cd4cc : SUCCESS in 27m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1193ff6827cc4ade9ad6ffef83082d9e : SUCCESS in 27m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eaad1fb8dc2b4590b9dfa0b388ecc042 : SUCCESS in 17m 31s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e282bdb03bbf4b1f88e40dcb4ed11ee2 : SUCCESS in 13m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bafc8ecfcf4540d9b12bd7e0da0dcfe5 : SUCCESS in 17m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a0a48f6b66884fdd907b0033cd62e5f5 : SUCCESS in 5m 25s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bf2778c0db44455c9628e59f1c957832 : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a01e9c02c1a40d49c83e9754ff641c3 : SUCCESS in 7m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6187b1e271a04cb486eeac7c998fa2b0 : SUCCESS in 10m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/35a1fac0a60e4269804f5b2b30f6c8c1 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/613d3b50e3f444b0ba802990b57b291f : SUCCESS in 48m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a6c0e6131114595b41ac400a3ea7d56 : SUCCESS in 1h 10m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f66df73e29f94c439fc4d86d5054ab77 : SUCCESS in 1h 16m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/3be2fd7a95f644baa8227eb99573be6d : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"67fd6d986f3c6836d3b349666b63d5ada8d0c1e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 00:40:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78de2d83353c7c6f4a1e51d85e09c6c053328ac6","revisions":{"78de2d83353c7c6f4a1e51d85e09c6c053328ac6":{"kind":"REWORK","_number":1,"created":"2021-09-06 14:58:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/95/807595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/807595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/807595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/807595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/807595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/807595/1"}}},"commit":{"parents":[{"commit":"c8b0e9c71f9dccb71fe6ceaeaad8633bb8af0160","subject":"Merge \"update constraint for os-brick to new release 3.0.7\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8b0e9c71f9dccb71fe6ceaeaad8633bb8af0160"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-06 14:58:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-06 14:58:26.000000000","tz":0},"subject":"update constraint for oslo.db to new release 8.1.1","message":"update constraint for oslo.db to new release 8.1.1\n\nmeta: version: 8.1.1\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I5392904b0e34b0cef9c7833d3801c820779211ee\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I6e2570a231bd4d5d542709fa857c7d7da2024a25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78de2d83353c7c6f4a1e51d85e09c6c053328ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78de2d83353c7c6f4a1e51d85e09c6c053328ac6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
