)]}'
{"id":"openstack%2Frequirements~808170","triplet_id":"openstack%2Frequirements~master~I7b453bfcd8486909d85067e6035b1a845e82c5b1","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I7b453bfcd8486909d85067e6035b1a845e82c5b1","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-10 06:34:19.000000000","updated":"2021-09-13 00:57:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1396a2b7a247fd0b2d5588659ea2afc06183f71","_number":808170,"virtual_id_number":808170,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-12 07:56:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 06:48:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-10 08:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b28895c8cce66ad9892ffe506961bc1a27c9ae6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-10 06:34:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1742eaf47e2efb464baf749bd3a2eb01515ba198","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 06:48:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ab19752da5fe4e6a959edee89e38ab7e : FAILURE in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"29cd2c4aaf8538a92e4ee293862637676c2272de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 08:44:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69db105938bb416797296078bcd26eb9 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f981b2f8c9b409b8a20ff6b5032632f : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b476afde5db44665942bef320462cd88 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dc86a6e2f214929a3fc5385e69eb814 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e71c6cd7e3404aeab3403d07848ee023 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c84f87544c7c472d87b1bde4d0f047b5 : FAILURE in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3aaa15b3dde4c1b92af0c204d7e07c9 : SUCCESS in 5m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a6c8013bc054ef28abd9ec7133c62bf : FAILURE in 4m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c2cf06b100304bb68873c7f9bc08ab32 : FAILURE in 5m 23s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d71455efa7af407180f34e2814db1b75 : SUCCESS in 7m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/40e1c68d3ddc44d2a1bef826e27298bf : SUCCESS in 10m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e9542eba186e41aeae105b2ad1c5a9e5 : SUCCESS in 6m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e966bbccb8b9437f8ca10d6e6a1e702e : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09c1256b7fee48afb9ff94a7542a005c : FAILURE in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7b7a685e31cd4157a5377f7f214131f2 : SUCCESS in 10m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8abb2acc386b439dbfdd8866bde0de88 : SUCCESS in 16m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c5ad4c4212474c6ba23c66658e7bff54 : SUCCESS in 11m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f6ab30dfe02e4ecf909437ad611269a9 : SUCCESS in 6m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/28acc904b2f1459487d3d6648272933d : SUCCESS in 24m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/243ab3a75a2c45849423233c02f80770 : FAILURE in 13m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b32b0bd465454fb0a25f51fb46fba73e : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5402bb99eaf34b6b8ba9ca0743bdffe1 : SUCCESS in 18m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7962349072e84c738adfe44262610889 : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c4eb9b70c5c4604b4930938c89b0a8c : SUCCESS in 8m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a8cb5cdfd87e49148e9d83771bd9dad3 : FAILURE in 16m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9127df47bfc045d3b6f9499dc2caae7f : SUCCESS in 45m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f72dee294cb544ad8d794054200d417c : SUCCESS in 1h 22m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cdfad8aa3626c225d910c948fd8e8259e934c060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-11 06:37:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44acdfb7763d472184c54892a6797f98cf7cbd5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 06:49:45.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b6c6d142a43541cebc1f9fffe3f1b6eb : FAILURE in 10m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b8643a94555346a100e2cd6a1736e45607b630b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 08:26:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3ef196a93643c4a4b7c2bb246f209b : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f378e09b964102a931a1f172274d83 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afbbbd099d0c439ba99176821bfa1dbd : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cad1b29ba7514e26a8fb5e711d133e34 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7db93e665ed64bd285f89ff904731953 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02af0bcbfb59436baa6b77899250906f : FAILURE in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7bd0b342ef841f0ac9d7dee8bf6c486 : SUCCESS in 7m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de45b4cadbbe439fab8f83bc9c120ccd : FAILURE in 5m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db001ba2231446848a54cd50375231d2 : FAILURE in 6m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/06c335b9797b4966899ebb42c177c137 : SUCCESS in 7m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4d5e1503f51c486088d004b1767fd998 : SUCCESS in 11m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6f088ac718e84d59968ae19d695bbb4c : SUCCESS in 7m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/77c3eccf2b6e46af9325e030515d5fdd : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7067098bda1c40d1be27347ea735a5b1 : FAILURE in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a3ea0fdd35d745d9b2874630a03fa8d5 : SUCCESS in 10m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/12c054d1a2c34328a2ce2fbd3c487bd6 : SUCCESS in 14m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d1eede562a344bf690e447e6945285fd : SUCCESS in 11m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/106cfaca02aa4c6da1785bd2846b1a4a : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e3b09727795b4f19b6117c134f5576a4 : SUCCESS in 24m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a6d65db479114450b9df6400a922d466 : FAILURE in 11m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ee054a68caf04faeb93719d36480554e : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3557f309d0474f05bd3e55e11f7e4bc0 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7fb62614b7f3450594783c75ff72c303 : SUCCESS in 17m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/131372c48897493c867a2bf3b4cf5b99 : SUCCESS in 6m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cb676ddcd07046dbaedb17bb9e83fce6 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49256d19e8cf468b834beb9cd8263eeb : SUCCESS in 7m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3c2f24217b7c4b8fa96f295a48044147 : FAILURE in 15m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/20280c68a40d4e22815bbb66ac1855ba : SUCCESS in 43m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/537ae8639d874a95947f32fa6f192cff : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":2},{"id":"855d28a501eca24c85fa7a747f859762da6e9fa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-12 06:33:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6c1713878d08593128bf10f3915998d92fba62c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-12 06:49:46.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8df0e7e7979c4fb296903aef8f78efd2 : FAILURE in 13m 26s","accounts_in_message":[],"_revision_number":3},{"id":"8b852ed1d0d1a37c86d6246eef47d9cabef9c293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-12 07:56:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6297aa9afab4636a3197ab68be52987 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de4606a3b29408999586fd2f4608522 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ef7b80ddf224d60a7aa788735c34e8d : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc173d7cadd74bd1b68256e82076930c : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f85a7f7ad1c48a28fd5f6c941f51a6c : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13f4f470949f4387a78853dc0423d439 : FAILURE in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/10009db55b064661a2d56148fa9905d2 : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f7d9f77a49b46d3888a8c4de1a2a53d : FAILURE in 5m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b66679abd7004fb48684e0e3150b8aa9 : FAILURE in 6m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b6f69df3c9f44eddb1d77661ae678d7b : SUCCESS in 7m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c30fe230ecb045d98c1d005bff0ab75c : SUCCESS in 9m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/26e0d76a3e964efca0e97242c51d0ca5 : SUCCESS in 6m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f9cbdb3efbe5414686f0c77a8f9d65d7 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eba7d55654704dcf81326dfcbbc461be : FAILURE in 9m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c5a8e297f38e4629a0dc57994913f2aa : SUCCESS in 9m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/19b091d3e85348de9baff86f86bd0de8 : SUCCESS in 15m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c4bdd56297fc4bb5a2bc0f324250c879 : SUCCESS in 11m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5b3d1c2480694b648b29ad75884c6740 : SUCCESS in 6m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7c3e6e8d8ce142dbae28d02217c3312f : SUCCESS in 26m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c702a0bad3284486aa363a4c0d692df3 : FAILURE in 12m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e8672efe9910427d9a6814792400b4d8 : SUCCESS in 6m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8565932ca3b44630b80ced5d1be7b76e : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d61e710e0d0422cb7abf59612fe3177 : FAILURE in 18m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f4a9c489f623423fbd34221dedb080c6 : SUCCESS in 7m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cf6303216fae4fd895cd000be0368978 : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/998d94b8b8a2486887aa22011895e62c : SUCCESS in 7m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/962f1ad62cb948bab92b071249b1ff8a : FAILURE in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/231455a5778d4734918dea4b5e2d10ec : SUCCESS in 45m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53412fd0dd5d4f47b7fee134f8f9a490 : SUCCESS in 1h 15m 26s","accounts_in_message":[],"_revision_number":3},{"id":"e1396a2b7a247fd0b2d5588659ea2afc06183f71","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-13 00:57:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"44a9075fabf56bb949f4d15c59fc29a32a7db2af","revisions":{"07df2d8e10f767a3f40afd497c8826297ad2eb52":{"kind":"REWORK","_number":1,"created":"2021-09-10 06:34:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/808170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/808170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/808170/1"}}},"commit":{"parents":[{"commit":"a8d0553e6569bfa8557f43e4bb30b34c6f99610c","subject":"update constraint for oslo.concurrency to new release 4.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8d0553e6569bfa8557f43e4bb30b34c6f99610c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-10 06:34:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-10 06:34:15.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b453bfcd8486909d85067e6035b1a845e82c5b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07df2d8e10f767a3f40afd497c8826297ad2eb52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/07df2d8e10f767a3f40afd497c8826297ad2eb52"}]},"branch":"refs/heads/master"},"b897c1461f0c1b6e8ddc82592843735e1aec3118":{"kind":"REWORK","_number":2,"created":"2021-09-11 06:37:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/808170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/808170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/808170/2"}}},"commit":{"parents":[{"commit":"91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3","subject":"Add cross-placement jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-11 06:37:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-11 06:37:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b453bfcd8486909d85067e6035b1a845e82c5b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b897c1461f0c1b6e8ddc82592843735e1aec3118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b897c1461f0c1b6e8ddc82592843735e1aec3118"}]},"branch":"refs/heads/master"},"44a9075fabf56bb949f4d15c59fc29a32a7db2af":{"kind":"REWORK","_number":3,"created":"2021-09-12 06:33:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/808170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/808170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/808170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/808170/3"}}},"commit":{"parents":[{"commit":"91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3","subject":"Add cross-placement jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-12 06:33:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-12 06:33:42.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b453bfcd8486909d85067e6035b1a845e82c5b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44a9075fabf56bb949f4d15c59fc29a32a7db2af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44a9075fabf56bb949f4d15c59fc29a32a7db2af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
