)]}'
{"id":"openstack%2Frequirements~808409","triplet_id":"openstack%2Frequirements~master~I2354b7d455168d8f03eacfe22061969f7961c991","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2354b7d455168d8f03eacfe22061969f7961c991","subject":"update constraint for os-testr to new release 2.0.1","status":"MERGED","created":"2021-09-10 14:57:46.000000000","updated":"2021-09-10 21:10:45.000000000","submitted":"2021-09-10 21:09:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808409-new-release","meta_rev_id":"e8ab74305b5aa444425ac0d1a9916ed39715e9c0","_number":808409,"virtual_id_number":808409,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 21:09:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-10 18:30:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-10 14:59:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-10 18:30:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 14:59:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-10 15:34:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-10 18:00:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 18:30:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dbb5822869bb4a11c844268b76f9f298cfd8f0b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-10 14:57:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f3b2dad007be40815bd5a1ed44bc5975b111424","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-10 14:59:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7057acc9ac799d8bb1a1dc4eeab33a4c92c49d55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:34:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6f2cc4fc55924521801ccf23f9ce25a3 : FAILURE in 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"84c9e5f008fe46d036b4e6732f4f9941dcf21dce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:00:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb8caca1647490c9e591d9d6d5aca4b : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a3d8f681a1428092a62aca1e2bbbab : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2282c1b3c3b14f6ea414e6cabb401a2c : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00607af172804d3980e50aa9301e984f : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f67c4c6211a44832aaf167366501ce1e : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6f1d4249c3940c5af1ebe7eee78101a : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/81d8df6b2daf45d498e077440a27c964 : SUCCESS in 10m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d862706cf6864c4d902725e116f74886 : SUCCESS in 13m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/33e85c5d77bc4daf910bea8630f24570 : SUCCESS in 12m 10s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/052e6c56fdcd4f45895cb7ca9f9262e0 : SUCCESS in 10m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1776fd9435d544308c31c3e66590688a : SUCCESS in 12m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9a0ad8c8558d41e6b33c4d0dd16f7e16 : SUCCESS in 7m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/693becafc9014783b502bf29fe3a2f5b : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8f4fe291d2b641a28dd8a031fbecb46c : SUCCESS in 11m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/312087cfbed641b69be9559510d2c961 : SUCCESS in 12m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/be8785a9733b4d55a8cdd1b46d467515 : SUCCESS in 16m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3faf45df5b7047fcbbfb8afc9828442a : SUCCESS in 20m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ffe2449c9e7f4eb6b193208fbec446ce : SUCCESS in 6m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5d001aabf87e446aaba3773cff5e57af : SUCCESS in 25m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3d1e4c2dacf64b808cf7ba27ff69228a : SUCCESS in 17m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0f7ae67a3ee4f998840be53d54e07a6 : SUCCESS in 5m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11ee2c0341f04dc088d29f1ffe079002 : SUCCESS in 21m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/45da4b96ef70427fb841a7a1f6e20d21 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/66dc870cba664e1d8b02a4989c68064a : SUCCESS in 12m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b8cf62ec6a2c4a9dbb13b81d88b4448d : SUCCESS in 14m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de10f9da316144b4a9153aa144ab2938 : SUCCESS in 51m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ac14bda98ee4233b41cb0582d6b40a7 : SUCCESS in 1h 20m 10s","accounts_in_message":[],"_revision_number":1},{"id":"665e00988746408c1aeef8c3c1d8f6a40ca52d97","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-10 18:30:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d5340f89639303d25824576fc525dcc219a062a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:31:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aee6f8182d4afa593337c0ba0d77735913cbb8c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:09:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337441e0d6124f4f9be55e0268ed1a9c : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/851a8c8d2e1a406a94eecbc2582693ce : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45eb5b631b4949edb61392210d08da57 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d67127ad34642bfa6dcb344d0b97a99 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d495312430346a19b0b8bee9220b5ef : SUCCESS in 7m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61df5365edec466292e3660ca4f67f54 : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d31e6eb9ad3b48f794bc17e299ade93c : SUCCESS in 14m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc6baa89b08f4b1195c6a505cba7b5a6 : SUCCESS in 14m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fea0e9940c1f44da81adc796da88e1a7 : SUCCESS in 13m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2c8b7f5ac56c4378aac3cbc3abe40178 : SUCCESS in 8m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c0c1af5c28ca4a23aec2d61f4b58452f : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dcfff895804348c4bae1f1038907b3f4 : SUCCESS in 10m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f529f7ad3d354067a7b04dbc8e7b632a : SUCCESS in 10m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cfb6b8b59d3d41bfa4630239c52dabc8 : SUCCESS in 15m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ec2ef24d37ab4e3fb21fc33b2dada294 : SUCCESS in 13m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/61f417ae5c044a3bab6135a16440af5d : SUCCESS in 7m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1298f6e8e1834e4ca51604891fd562f6 : SUCCESS in 26m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3eb42c4b68394d8982338065da32d5c9 : SUCCESS in 13m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a438a27120c40e88667bd5f68cec97b : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbf959eff7844ec581eb25ca1cd91810 : SUCCESS in 18m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a798e43f7df04ab98541c3a726b3bc6c : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c64c7ade92864cddb8e376611cc062f4 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eb4690b8351e4102a6e2e95be0364fbd : SUCCESS in 16m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0ca3468882d445aaf4d5518bcccfa63 : SUCCESS in 43m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08569fb8f8654d06b0af0abd0406bb65 : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7f23ddc0309239d5cadecc53e74bf01ab517a929","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:09:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e8ab74305b5aa444425ac0d1a9916ed39715e9c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:10:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4db3e7c5a02b4a1793927c4b5968f376 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2677620f0b9a7d1c6b5f41850249b5459390de20","revisions":{"2677620f0b9a7d1c6b5f41850249b5459390de20":{"kind":"REWORK","_number":1,"created":"2021-09-10 14:57:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/808409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/808409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/808409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/808409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/808409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/808409/1"}}},"commit":{"parents":[{"commit":"a8d0553e6569bfa8557f43e4bb30b34c6f99610c","subject":"update constraint for oslo.concurrency to new release 4.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8d0553e6569bfa8557f43e4bb30b34c6f99610c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-10 14:57:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-10 14:57:41.000000000","tz":0},"subject":"update constraint for os-testr to new release 2.0.1","message":"update constraint for os-testr to new release 2.0.1\n\nmeta: version: 2.0.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Martin Kopec \u003cmkopec@redhat.com\u003e\nmeta: release:Commit: Martin Kopec \u003cmkopec@redhat.com\u003e\nmeta: release:Change-Id: Id053729ca83d5488377a5a2d982a9a88fd77be58\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2354b7d455168d8f03eacfe22061969f7961c991\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2677620f0b9a7d1c6b5f41850249b5459390de20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2677620f0b9a7d1c6b5f41850249b5459390de20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
