)]}'
{"id":"openstack%2Frequirements~808415","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I857d73cd60ac06e632cc1b7a5d5000c640869b22","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I857d73cd60ac06e632cc1b7a5d5000c640869b22","subject":"update constraint for neutron to new release 17.2.1","status":"MERGED","created":"2021-09-10 15:12:58.000000000","updated":"2021-09-10 21:09:22.000000000","submitted":"2021-09-10 21:09:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808415-new-release","meta_rev_id":"2238929aa87bc2909f2dbcc98636e2c1c01d62e9","_number":808415,"virtual_id_number":808415,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 21:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-10 18:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-10 15:21:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-10 18:43:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 15:21:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-10 15:37:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-10 18:32:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 18:43:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"40418683e53bd865d5d30f3f32463fe6e10dd79a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-10 15:12:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66728c04f3b14f3bfd195e45e1818a053699b268","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-10 15:21:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e89c0061f03f61ab6ab053a6b31805536d1cd593","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:37:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a18eece2fa7e4f6e8346728d9192f97f : SUCCESS in 17m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e227b6da7bd641fc61ce24df4bea2c4f8b261276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:32:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4d6b25d8da4a9e8b3b5c61636902d2 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c9630b1979f4ea7b258201dc33d2c85 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e74d738f44134bff836cce858eed556a : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d95016d79993437982618fba20118a83 : SUCCESS in 8m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6e8b244e8784d78931cb54d6b23641f : SUCCESS in 6m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cadaefc9b9cc4e2ba3881fa7198b60e7 : SUCCESS in 9m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/84876939599740b28762583f28b6caf2 : SUCCESS in 14m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dabcdea26ce942d59cd7b115fe089a71 : SUCCESS in 14m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2da5672b5d19458dba42140010614a49 : SUCCESS in 7m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fa0cc2d921c94ec18a8e3cae3036ffb8 : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b82979b12574a63a18d3c04931f4aa3 : SUCCESS in 11m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4b2762ddf8244e8da4cc22c805c389c2 : SUCCESS in 8m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e0cbf5e7a3d41ac8fcbd19b16711077 : SUCCESS in 17m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8009dc3566b4f818f9ba8a1d7078060 : SUCCESS in 27m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/268baa259edd45bdb6725d394fa65235 : SUCCESS in 17m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/872967508ffe46b38b469b519315667b : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df2d5ba06dbd41ef941ab708635eb2f0 : SUCCESS in 17m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b69216b3731a4ab4affac530fce76502 : SUCCESS in 7m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f3eaf9ab37b44295ad6e7a35a7dc84ae : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/daa98e916108485387b9894e36204fab : SUCCESS in 12m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acdedc74627d4560a7b6d0ace89d5275 : SUCCESS in 57m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1bbb81ef283346d6bb9d814cf61ca3f1 : SUCCESS in 1h 45m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c5cd2696d7e98309c924c10ffcf51be6777035b2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-10 18:43:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1df7f1c312d106b33ec407813a18c4f124123d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:45:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"735c9103dd70a72cac3d813dd19441436332cc3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:09:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f160aa0eaa43bba90712e9d1d85e58 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f66bae25eff64a1db235e831bcc5a9ae : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae891946facd4fa29c11444aaa767b2b : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4a916078b9941c784dc2db625bf27f5 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f336d3d0cc474fe683221be8bee231ca : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6d10c57246a4c0f854b0c9d34289201 : SUCCESS in 11m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ebf6eb9b1b1426784def1d40d473191 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b4c7580038994f0cb84bc241c904c89e : SUCCESS in 9m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4aec15a1c2ff47349d570cce2c79ee59 : SUCCESS in 6m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/50e07447560e4b92b68c5dd498327c8a : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8914e289f7a748dfa1db4456a6dbd2bb : SUCCESS in 10m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a9d51ece9854944b4e1e07cb46d699f : SUCCESS in 8m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/73b4cb71af7040618ffc2b30ed1aff46 : SUCCESS in 12m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/20b23280270f4bea94d03aa5c1f0c897 : SUCCESS in 21m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0986d7b3a56a4cbfa4cb0bb222895755 : SUCCESS in 14m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea4fc1bb8c554428b962ebb1719f57ca : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/17b65b70d12f4d30b10023a239f4c730 : SUCCESS in 17m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6c440f45ab834e50aec3b3b5b4ec720a : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7bc220bb55574d619aba0d15b8f16fe4 : SUCCESS in 8m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ccf9708b09864be897a780b08c7d1580 : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4854978a51444a1ba11c7bbefab62a01 : SUCCESS in 48m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d97b2aafb76f4fd7af52847e99ac6ccc : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2238929aa87bc2909f2dbcc98636e2c1c01d62e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:09:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a02a46c497aaf86701ddf94e20660e2bf7f8b4e9","revisions":{"a02a46c497aaf86701ddf94e20660e2bf7f8b4e9":{"kind":"REWORK","_number":1,"created":"2021-09-10 15:12:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/808415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/808415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/808415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/808415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/808415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/808415/1"}}},"commit":{"parents":[{"commit":"9cb8b4ef9f3dce53f76ed6c6706c990fd4dcc2f5","subject":"update constraint for os_vif to new release 2.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9cb8b4ef9f3dce53f76ed6c6706c990fd4dcc2f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-10 15:12:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-10 15:12:56.000000000","tz":0},"subject":"update constraint for neutron to new release 17.2.1","message":"update constraint for neutron to new release 17.2.1\n\nmeta: version: 17.2.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: I44b055ad48795d9ad9674ab282f50fedf4b59ece\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I857d73cd60ac06e632cc1b7a5d5000c640869b22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a02a46c497aaf86701ddf94e20660e2bf7f8b4e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a02a46c497aaf86701ddf94e20660e2bf7f8b4e9"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
