)]}'
{"id":"openstack%2Frequirements~808463","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I3c48a7292fa63c304ea694110ae9ac81267193b6","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I3c48a7292fa63c304ea694110ae9ac81267193b6","subject":"update constraint for neutron to new release 18.1.1","status":"MERGED","created":"2021-09-10 15:22:55.000000000","updated":"2021-09-10 21:09:18.000000000","submitted":"2021-09-10 21:09:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808463-new-release","meta_rev_id":"dd3ab0bf4b06e64eea043d83795b181e123001a9","_number":808463,"virtual_id_number":808463,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 21:09:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-10 18:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-10 18:30:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 15:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-10 18:29:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 18:30:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"91d32ecfcea1f42ee17e00cbec4dabed9219154f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-10 15:22:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5f7bd39a992de9984634e7b9f75ceb735b23a41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:54:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d53f246b50704564bd4663714f3ba6e5 : FAILURE in 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"752e4a44c1bd886b431f01d8800dd60701cdfcce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:29:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e6836e33ff34373a0854af0a9389607 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3235c794a7074f528948ca721b884182 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ef0882029ab4952a00d16aa5a4e538b : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08d73c35af0c4ac5a1833916f849a789 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc283a0c8c6c434abcd84bbc108e45f3 : SUCCESS in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c608edb570d4925a8f061818e021724 : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a99e7faa9503447b8eae70c74abea298 : SUCCESS in 16m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1b503f6c12dd40c99dc09cfdbad7441f : SUCCESS in 12m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7951ec5ccef6459d89cfafd6c8707bec : SUCCESS in 13m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4fb1968d5da4d1b818be257ef472902 : SUCCESS in 10m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f51c272acbb1407b85c046d6464eddd9 : SUCCESS in 12m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ad66c3aca3945e5bee31f4beec7a2e7 : SUCCESS in 10m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a9322d8b88da496d81a2835a6563f17f : SUCCESS in 11m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/39b26098a2d4458b9aa7d9fefbf175f4 : SUCCESS in 15m 15s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c3f51a47ea274043b206992dd792075e : SUCCESS in 6m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/adc83ed19cd145119a35eeb47d2463b0 : SUCCESS in 25m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/58b9fad73ee444578b72c71388b5cc0d : SUCCESS in 15m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ecee1567554498ebe8ecc3319b43067 : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/399505c2b8c1484fa7e8c45b59b8b22d : SUCCESS in 20m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9f09740707d740adb0fae916732abadd : SUCCESS in 7m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/799913aec9b24fe0b16500590796f35d : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9110bba0b2294e46a3b0af3f2bb54b86 : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b8dfca6368d44a08781b74692924697 : SUCCESS in 54m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b867c69110934c6da27668bff373a3b0 : SUCCESS in 1h 28m 18s","accounts_in_message":[],"_revision_number":1},{"id":"7724a4a2156b06f875b8303c1a4497462e5a51d0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-10 18:30:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bca8d266ad207eaa0f21a61c057ce11d94988b5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:31:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f02431e366fa7c0a8399d2abcdd4eb54d39c2a59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:09:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fcaedefb4864e91a90f548aa757e686 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2fce87e65324e1a8a6685b29fac376c : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50a9584338e742a396a4ed66d0171de3 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95d9a91f8d8d49e8a9f705d234104fa4 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/addad67353794155a9b491c02265e622 : SUCCESS in 5m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a171d6703aa4393af9b79ed4709ba66 : SUCCESS in 10m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7600b834bb3b40a589f698b858818765 : SUCCESS in 10m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/866c07a472e24a188d2dcf22aa42541d : SUCCESS in 10m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/534d236964d14ca6b43db59d3fd693ce : SUCCESS in 6m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3034239f18bf4e188608088876ce8bb8 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6217462581244e1da6bc305da7db29f9 : SUCCESS in 14m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a2d85db433eb4cedaceea6aa77a340bf : SUCCESS in 10m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/46da98a3a52c4fb18c9a806f5fd6607d : SUCCESS in 13m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/28399d4b33634741891bbc8c51f00914 : SUCCESS in 6m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d4d6e5d7003479ead37c2d4ac03b5cc : SUCCESS in 25m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cbcebc9ec4c84deb9a8afb357e753fd5 : SUCCESS in 11m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4588a0520c14a2186e188e13801e9ac : SUCCESS in 8m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95a37ef04d5d44a3b8291f4f4beeecd5 : SUCCESS in 20m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b669955390124ecfa4d0de16dca35364 : SUCCESS in 6m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6347aee0960d4d18a68793883eca1a11 : SUCCESS in 8m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef05add4e7ee4525beaa7d1aee17cb5f : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef649db665af4ef0b2b3c5c6573386aa : SUCCESS in 49m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f16fd34dd9c145c4baf08d009c449196 : SUCCESS in 1h 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"dd3ab0bf4b06e64eea043d83795b181e123001a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:09:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa069d105884ec24eae49142a092b5a2af4b6f82","revisions":{"fa069d105884ec24eae49142a092b5a2af4b6f82":{"kind":"REWORK","_number":1,"created":"2021-09-10 15:22:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/808463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/808463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/808463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/808463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/808463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/808463/1"}}},"commit":{"parents":[{"commit":"755ebae8af8b821c5072a06d026bbdad4e643b49","subject":"Merge \"update constraint for castellan to new release 3.7.2\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/755ebae8af8b821c5072a06d026bbdad4e643b49"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-10 15:22:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-10 15:22:53.000000000","tz":0},"subject":"update constraint for neutron to new release 18.1.1","message":"update constraint for neutron to new release 18.1.1\n\nmeta: version: 18.1.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Change-Id: I44b055ad48795d9ad9674ab282f50fedf4b59ece\nmeta: release:Code-Review+1: Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3c48a7292fa63c304ea694110ae9ac81267193b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa069d105884ec24eae49142a092b5a2af4b6f82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa069d105884ec24eae49142a092b5a2af4b6f82"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
