)]}'
{"id":"openstack%2Frequirements~808520","triplet_id":"openstack%2Frequirements~master~I625b2345248f2198ea264e7cafbe6e93e4ba4a1e","project":"openstack/requirements","branch":"master","topic":"story/2009209","hashtags":[],"change_id":"I625b2345248f2198ea264e7cafbe6e93e4ba4a1e","subject":"Add cross-placement jobs","status":"MERGED","created":"2021-09-10 21:57:03.000000000","updated":"2021-09-11 02:10:01.000000000","submitted":"2021-09-11 02:06:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808520-story/2009209","meta_rev_id":"a975e14fe0c57d97bf2bd168f9b810a168d64551","_number":808520,"virtual_id_number":808520,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-11 02:06:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-10 23:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-10 23:33:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 23:25:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 23:33:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"86028f6d55abdb2fd278c874354f830b0b39cd42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-09-10 21:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"362fee380b5a0e2caa1581cdc36c57ef78adcaa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 23:25:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a458c1cb6c99406d951b7170fa0e451d : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ec375ae3328449488155cf058fdf71e : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b2491ef7d6841f69619308edf8211cf : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a382e9f3cd8e4758acbbd3dc9eb1ecc2 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9426d8128634a60aa2f813bd6d5dff5 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbbb6be9d36c4720a54c44b3f2fcf2c6 : SUCCESS in 4m 39s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/26525eee672447e697d5b6862c4c1366 : FAILURE in 4m 29s (non-voting)\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1ad8ce06f5544b15b6c4131e4e7a7b63 : SUCCESS in 6m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3ad7516a63f5472c909ff72d7eab69b2 : SUCCESS in 9m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6a79555cb18f4e2289ee846808f52757 : SUCCESS in 6m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d13dbf2f90824eb6a24edc704e6ed342 : SUCCESS in 14m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a61fc1febd894b30932e8f6bb96483a5 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a8f77527145646a498fd37010bd98a77 : SUCCESS in 8m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2f993e4cecbc40cfae583d46ec64e949 : SUCCESS in 14m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fc53fbd2fdf4404fa6cd51858d40ab0d : SUCCESS in 19m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bc6653de0aac49e788b57e542ecf80a3 : SUCCESS in 6m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8058d11b9fb945f7b4ab4c594dd93f0e : SUCCESS in 25m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a8f5964987f74160ac75c9a1ee22f96a : SUCCESS in 12m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/667162f4e8844ab5955cc0d9375b30ba : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5dc6ecab87054181b6ce8a71e1d7059c : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/778c962876ed4b53a07a793366a683d5 : SUCCESS in 18m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6967c76fc67349d99c5e9279dc8071d8 : SUCCESS in 6m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e9f235ffe80e49ba8a4ace61a626bdb5 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f11900de4e884b9281f80fd0d2657092 : SUCCESS in 7m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7f2a5534cb294022bde84da678805a07 : SUCCESS in 13m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/189a0ce805304e5bbf829425b8dd912d : SUCCESS in 46m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cff7bf4975f5428193af28bb521cd174 : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d5839243a7d444a4baa6dc8ba2b79ff752ebf981","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-10 23:33:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c5bd5ff84392a9c3f1df1dbff65b51dd194a876","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-10 23:33:49.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2d3d0ef4a1a09ead251ec0b634e98b7af573c2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 23:34:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d44a12f90dd254817186fc8b60c4d53bc8ab0f9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 02:06:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a987d567df411d9203aa6832e79d59 : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0aef0d7799e415cac4ec74a456a01c6 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7785a2de2dd24ae6908ba4ae75606b68 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/800051a0f778445ca8315d7cfb28a585 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5421804dd0c4ab0b02f43597cc7ba28 : SUCCESS in 5m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/385e50a837a44c3ca987402278eddb56 : SUCCESS in 11m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d2f935a2286a45069a2f34d820cb993d : SUCCESS in 6m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/85be0e951b394a4f83d42a8b959e2be0 : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18ab6431547948e3a4f789c23f5c8844 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b8616fd92763423fb23e6be33da14187 : SUCCESS in 8m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/28ef7fded308486e973bc753753e0547 : SUCCESS in 19m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eb28514aee6f4aa3b74ddd845458e8a9 : SUCCESS in 12m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4981292bea794ea8a94593b03ff7f012 : SUCCESS in 6m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8ae9465e2ce54245858e9520dbd5a3c3 : SUCCESS in 21m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5549f7dd16c049c490c8010120268c50 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b6a7c727cdf347b598cd4b7d6f885233 : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e17dd81aef2243ed89be40aed947908a : SUCCESS in 21m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/14a695a87d0f4398832fc9085744a10f : SUCCESS in 7m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f3b758b81fa14f6fad60c2a04a04b7e2 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c5ac82b0e2f04e199cd070b176d026b7 : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ace5de0843c14956b704aa802803d90c : SUCCESS in 14m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a22203b70b5470a9ea9de47aebae909 : SUCCESS in 43m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce79265cc99948aaab0cafbdfe7ca05f : SUCCESS in 1h 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d6b7f99525e2c61b8425dc554921763c4b4b30c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 02:06:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a975e14fe0c57d97bf2bd168f9b810a168d64551","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-11 02:10:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88211f42d5904c91a1e60b22d3e7b894 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3","revisions":{"91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3":{"kind":"REWORK","_number":1,"created":"2021-09-10 21:57:03.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/20/808520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/808520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/808520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/808520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/808520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/808520/1"}}},"commit":{"parents":[{"commit":"2677620f0b9a7d1c6b5f41850249b5459390de20","subject":"update constraint for os-testr to new release 2.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2677620f0b9a7d1c6b5f41850249b5459390de20"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-09-10 21:53:06.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-09-10 21:53:06.000000000","tz":0},"subject":"Add cross-placement jobs","message":"Add cross-placement jobs\n\nA recent requirements bump of oslo.db from 10.0.0 \u003d\u003e 11.0.0 [1] caused\nfailures in the placement project gate, in the functional tests.\n\nThis raised awareness that we have not had cross jobs checking with\nplacement since placement was split out from nova during the Train\ncycle.\n\nThis adds cross-placement jobs which correspond to the existing\ncross-nova ones.\n\nStory: 2009209\nTask: 43275\n\n[1] I03d79b19774650f3df7129157e6a12998fd33447\n\nChange-Id: I625b2345248f2198ea264e7cafbe6e93e4ba4a1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
