)]}'
{"id":"openstack%2Frequirements~808585","triplet_id":"openstack%2Frequirements~master~I139233382d1131f986d23e9dc7ed7ed92ee3a89f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I139233382d1131f986d23e9dc7ed7ed92ee3a89f","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-13 06:43:39.000000000","updated":"2021-09-14 15:12:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aaf98b04e285bc537f659e1d6406d1d1bb33eabf","_number":808585,"virtual_id_number":808585,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-14 08:04:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-13 07:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-13 08:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a71f55c0954f08590d7f1f5a070f355f2118f751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-13 06:43:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dbbabac4f6072f31e720ac9b5d0ecf777ebb759","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 07:37:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/96fd428b81bb4521aa1f5f82bacdbbc4 : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0e5bd8b67b09f5de1aa71beffec57a1bd080762c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 08:21:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc3ee41816f410486b297879ee71f09 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/310c31fb8d464b4480d2c9949311fa70 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/665ff928265f4b21b9cb1a42a73f1b44 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/331347c5ec9440f085a6c32e38b843ee : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/193dd58493384d4c81a4d092aa98c0d1 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54a2438737484311badbda22d2166c89 : FAILURE in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/261956b668a1481eab0507edfa68e5b4 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cea2a1c01a5d4ef7bfbd35be0480217d : FAILURE in 4m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ca89e25eafc4526949a82718e303699 : FAILURE in 5m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/293fa278092b48b3b896306f5dfecb10 : SUCCESS in 7m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/96833efcd45d447bbcc0d6042cde8687 : SUCCESS in 9m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ed2795cd57cd4f22b3b98a5003635d90 : SUCCESS in 7m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/56c3b12bf22b4d578f897dc9ec02c5b9 : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc0514b8e96042958d89fb60c369d2aa : FAILURE in 9m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/24fff40891b64531afd39aa14969bb42 : SUCCESS in 9m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f730c7356bf44828a34322ed80c84018 : SUCCESS in 14m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/17504614b7314ef39581612fdfd73837 : SUCCESS in 12m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/147d2f5ca13341e687276c75b83a65dd : SUCCESS in 7m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/736eb02666ed48a09833f1551592fe0a : SUCCESS in 27m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e8219b1f691d47a192b86b625c1ec05c : FAILURE in 14m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/63dc19448cd045e0a051b66d30bc50f5 : SUCCESS in 8m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46d6134c05194118b6beb8c22991c670 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca6bd48c6a4344f5ace73a42da32dc13 : FAILURE in 19m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ece1b47f05554088a99a2f4803fef097 : SUCCESS in 6m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b62d4f5d502e48a29b0a27e876c29984 : SUCCESS in 6m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cafd59cedd20455c80b6e8a8ee3dd8a2 : SUCCESS in 8m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0f1fe91495df448892d0c3aa146752a7 : FAILURE in 13m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/339bf2de7be44b8b99fb12498498bae0 : SUCCESS in 43m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33bc73a2333442ecbc69a39deb270e2a : SUCCESS in 1h 24m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3c8e906534cc5e335bb2db6bdd6fd2d2f477d233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-14 06:36:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8de06c0373813921bd5154caab55953e8cdeb094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 06:50:46.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/03d5bab8294d45f285e25e456c765845 : FAILURE in 13m 31s","accounts_in_message":[],"_revision_number":2},{"id":"be9bdaa4019f16d65fe2288a8ebae3c046ba1580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 08:04:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc6e5caf9d24abf93a61534c3f765e0 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5b763720084abeb355a434cd240520 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bc5737fa23144929a2650e687fca2f6 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78af8c00e5504797aef714b75c0c8e3d : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78b4b333bdfe4b95a16c2e15696fd76b : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84cd7499854340f2aa059d105d1d40fd : FAILURE in 7m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7431a2080b17459bbe320e7d2ae6f9c6 : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ec6b73d1e9be49ddbaa79897d2b01b00 : FAILURE in 7m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78c5382227b44fddbabf059b21f21f24 : FAILURE in 8m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9afee2fafb2141bbb8d3695bf31a3b49 : SUCCESS in 8m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3243343a5a054581b8b2d32682dcac58 : SUCCESS in 13m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d1c646e4445742d0a0dac02f95633985 : SUCCESS in 8m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/84bfc52b2d144ffba12948abd4093fab : SUCCESS in 13m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/411231b7b80d405584f670999ba8e200 : FAILURE in 9m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4813db073c154ba397aa973f603e8a87 : SUCCESS in 13m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/86c99addf62c41698e32c7703f29074a : SUCCESS in 20m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9e78a0ac85414fb3840d2d2a8443330d : SUCCESS in 13m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/59b20a96f5524125bea0ef93c07cfc36 : SUCCESS in 7m 44s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/864f49badf174bbb9431babcc052cb67 : SUCCESS in 28m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/af8692fe329c40d2bef30a73829d7401 : FAILURE in 15m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/147ab1a57dfe44458822f7c1048327c1 : SUCCESS in 9m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d7d5106130cc4c038d222506faa1a2be : SUCCESS in 6m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/296cb9743c71487f9b9023fe34c638b3 : FAILURE in 20m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/93197faa9dbb4012be07f93099a0ee49 : SUCCESS in 8m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/abe3d1fd53ce463bba649bda1efc9185 : SUCCESS in 8m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/669cd4a835094cb48c1a6c966a482fb2 : SUCCESS in 13m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/991d789b127844d7a4907da3594cff8a : FAILURE in 13m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bc9ea4cd33b40299d48f613e90efdf3 : SUCCESS in 43m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/558582f1c5274e35b47bf17e1804ae5e : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":2},{"id":"aaf98b04e285bc537f659e1d6406d1d1bb33eabf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-14 15:12:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2750b9bea79bcb3613a12864124fadb20affb512","revisions":{"0fed4ecbf7b334102356f93f437c363f89fba3cc":{"kind":"REWORK","_number":1,"created":"2021-09-13 06:43:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/808585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/808585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/808585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/808585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/808585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/808585/1"}}},"commit":{"parents":[{"commit":"91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3","subject":"Add cross-placement jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91eaa0cd7f75f5f1b93226e4bb2de64aa4fed2b3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-13 06:43:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-13 06:43:37.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I139233382d1131f986d23e9dc7ed7ed92ee3a89f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fed4ecbf7b334102356f93f437c363f89fba3cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0fed4ecbf7b334102356f93f437c363f89fba3cc"}]},"branch":"refs/heads/master"},"2750b9bea79bcb3613a12864124fadb20affb512":{"kind":"REWORK","_number":2,"created":"2021-09-14 06:36:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/808585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/808585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/808585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/808585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/808585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/808585/2"}}},"commit":{"parents":[{"commit":"7d440eb91773aa8a9f455a3f7c69db780e926679","subject":"update constraint for python-cyborgclient to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d440eb91773aa8a9f455a3f7c69db780e926679"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-14 06:35:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-14 06:35:58.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I139233382d1131f986d23e9dc7ed7ed92ee3a89f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2750b9bea79bcb3613a12864124fadb20affb512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2750b9bea79bcb3613a12864124fadb20affb512"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
