)]}'
{"id":"openstack%2Frequirements~809153","triplet_id":"openstack%2Frequirements~master~I7c22d7cc8092222a91a1da2bd65958bb3a9599eb","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I7c22d7cc8092222a91a1da2bd65958bb3a9599eb","subject":"[DNM] Test Django3.2 with horizon","status":"ABANDONED","created":"2021-09-15 08:04:35.000000000","updated":"2021-10-11 12:34:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8fc6c72c80a0b93df597c2d6e4518f04cffc5d5f","_number":809153,"virtual_id_number":809153,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-29 18:25:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-15 08:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-15 09:12:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44c86c7711c4f8c62120d8f39ce7e01b2ebaef34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-15 08:04:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3ce9f544030cc734b411bde80951415b063330d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-15 08:04:53.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"05db95b62a57a262a0def778ab25846f015450d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 08:18:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/66e041d4f29f4409a1e7e171ad243a0b : FAILURE in 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0f79dd786692a6b2f87564ef0d90ec1b319b7874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:12:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce2149034b84058bfbc624bb1c11d6f : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3686afc88a934b1584c8c60c20de6a57 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c736d53dda6146b0887dbbd1f5ccca7d : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/846fda895f784d3bbfee2a82e4bac1ec : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/029028679e5245c08606e5f8714a0911 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82a5f233bc2c44bea080a08327ae9c30 : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2305703c20984bfca0e98da74d12520d : SUCCESS in 5m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/beafb0b5dd0e487db70ed1e83e2e8d4c : FAILURE in 5m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46264f07f3db4010a173790ea021ff8b : FAILURE in 6m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ae6d129cc6146af8b978d020a8d5a6c : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/65517f9f4cc64a9a94482bccda84a11e : SUCCESS in 10m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c4af0e2d2b0146d5959a9932acb8a45e : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a47dcfb865e14621aa891016aff951ff : FAILURE in 4m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0df784e43a14eb590e7f32b1b9248e5 : FAILURE in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e49599c0fda348b9a88210dcf5e95b74 : SUCCESS in 9m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5112a9c2e23848b8a748446f2a81d472 : SUCCESS in 15m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/de559b98661b41fe8fc131deba0ba4c4 : SUCCESS in 12m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b0b01ec3925042dfb90c1baea7b7a96d : SUCCESS in 7m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/adc64124e7614374909baf69a8148b25 : SUCCESS in 27m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aa7849ed17584f4eb6b49d391f876387 : SUCCESS in 12m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0cfd9a36224f41feb45f075681fca4ec : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/84f601166758476784a60286e2e1b469 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3d37651b8af4c7fae77b592b6ec9883 : SUCCESS in 18m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bb063bcaf8194b49b6bc5833eb4dcd1d : SUCCESS in 6m 44s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/31fa53db650b48cfa7bfca6413390e46 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/353fe958755740ceb41ed5c88ce08788 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5d1d6c310db0437f8979f4d065f45fab : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d77d50f981f74e31b12301f977189bcb : SUCCESS in 46m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6660afdd5dfa47c1812a90e1f56b661f : FAILURE in 22m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5da79a8392c9ab0e47e82d615b5fe05038965038","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-15 09:37:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c9d22804f401abb36c397247048f035361e8309f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-15 09:37:54.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"25d084c4fa5fac23cd2a43d199f65db2122e56f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:55:28.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0356f7c4d21044ea91e8dba6cb978a39 : FAILURE in 12m 12s","accounts_in_message":[],"_revision_number":2},{"id":"56e2c06c3be0c289bdf72ef424fb58bb15522304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:31:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86a6417784e445029fd60627dec34d13 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6afd24eec21b4924b4e85c549d126edc : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ad471f48b694ab7b40db742d5724ff4 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a5f12bd6321443bb5acbfb5e9a21eac : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f8bb50c828347b18bacd87a0487b6c0 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6157ff26d7714b70b40aae6f589189b6 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/89689f56dd1b4f27acb7c8b87289c6f9 : SUCCESS in 5m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46e0aa39e0d742738f3c4f60c93711a5 : FAILURE in 7m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2d30ebcdfe9f48529f88404e187ac5f1 : FAILURE in 5m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bd7bbac8a617413eb5504277d750b18f : SUCCESS in 6m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/225627c5e8d445edb5efeb7d8c4f9747 : SUCCESS in 9m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e827eaf63c0f4f6d96e0f315f624a875 : SUCCESS in 6m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c1aa8d2c8ca04dfc8c2db4e23232736c : FAILURE in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/706e7daa9cd047b5967479cc9b4895fa : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/36592bef15b84e0aa7dc1cd0610d5996 : SUCCESS in 9m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/112c6eb28f3a4d92ab8f8b97cc8ec91a : SUCCESS in 14m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0aa1a669efeb4566b85bdfeb1b303b32 : SUCCESS in 9m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1e89624aa07d4182a256dc28d1b52739 : SUCCESS in 6m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e5599f0c985b462c83dd3ef7fa1b687f : SUCCESS in 24m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0f7a28aedc834e339b5840bac899d126 : SUCCESS in 11m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fff1a609e72a41d8bc4707cc45cd3dae : SUCCESS in 6m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/18873352460941e1a48be8276b39146b : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1df7624a6095451a88a2968eea422ca8 : SUCCESS in 18m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/96897193f7f94c1285edf45fc8df3138 : SUCCESS in 6m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e41dd065e0c74a6ea1d2f3589f373377 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0888e180722047de9f1a9f30bcf1366c : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9a5166e7db44472b87fd650b5706186a : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f781c9a62e984b9ea6997c081054e518 : SUCCESS in 47m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ae667b21dc04b9fbc4dd01e60966d8c : FAILURE in 27m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4c7da94f6321e0572ad011bb3910e830e23ca676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-24 09:49:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"690544f62e09cfa715ec3db66f71c57b2937a114","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 10:04:16.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f2e3f84074224bdea711528b1dda3e35 : FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":3},{"id":"590c450f5acac8c38b7787f87e3add90440358d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 11:26:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f5a7463960475e92e49ec465b0032d : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3abd36d2aa3a4987951ac98c29f1e068 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea708006d6ec41b1a3ac928ff3fde718 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc76c65134894b45a40b4a2c791d1168 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0c8046bce0a49bf858531d7080e0181 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d98aed73be444a2289fa2aa01aa85ec7 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/05048571147a45f383381b1711ccea0e : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5222b8bcb8804a0eb38fca986741fe88 : FAILURE in 7m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/10b8f81a3abf4cf18799f4d798810989 : FAILURE in 7m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3631dca1399f4c3685b23c743c214ce2 : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/46d1bb0d2c0e4a40befae75bcad8fb0e : SUCCESS in 9m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a49f5badcb344082be290a80e2aa144f : SUCCESS in 6m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/086d4e3b81284a6a969e8f3f8193dc09 : SUCCESS in 10m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2842665cc5fa4ad3949062139807736c : SUCCESS in 9m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fdf222aaec3440148d08251d982f627f : SUCCESS in 8m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f52ba1c79693451c80dcdcb82ca2a194 : SUCCESS in 15m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/81ce4b1389204c15b000936175fa15af : SUCCESS in 10m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/45c1642634d141cbabe0ea87a124df37 : SUCCESS in 5m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/19610b3c8fbd403c9f79387000287de2 : SUCCESS in 22m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9b3d7f0103a740f38f1f73ba90c76b59 : SUCCESS in 13m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ba73ac8c556e4c3f9f1438b6e2034c99 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4301ba7dcf8f460b83284b4505c6805a : SUCCESS in 4m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20d24a11da8a40a8979c619b70027f80 : SUCCESS in 20m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7131fc599f2d45708cd425b649374247 : SUCCESS in 7m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7303a89c636a4d3bbff417637e27231e : SUCCESS in 7m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82737d2646a8478b94d986886cefd678 : SUCCESS in 8m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c6b68969ffdf4cdea9cd6b1626282059 : SUCCESS in 14m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23512f0848354705a2afef9d60082e39 : SUCCESS in 53m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2c847d9347a4dd699ef4a55901da36b : SUCCESS in 1h 32m 10s","accounts_in_message":[],"_revision_number":3},{"id":"324921e6432e291b9f41259bff2123368bcd96b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-28 13:08:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4090b05149cdc5347a849c3d0a391e1ddb5736fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 13:28:17.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/365ed681595f4792b1dc2a21dc0d4810 : FAILURE in 9m 35s","accounts_in_message":[],"_revision_number":4},{"id":"62aea2805fe6c40b4827cdfb24d7de67b273a8ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 13:43:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ab9299d70e49648f877f9b2a7ece65 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa128b274bac46df80aac34644ecffe8 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a5e6a70ebfc421388c28327c0267809 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3304a3c4dc714879802757e488468ceb : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d32cf7a3948446f2a92df1288702ba77 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a22c323ba7664510938dd81aca5cc228 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04e1adf1e8dd4d93bbe2f765925fabc5 : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/738ebb33f3f04fe09812e3c2ef07f7b3 : FAILURE in 7m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/34baa2e412fb4b8db42e1f07b72dc52b : FAILURE in 6m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/03c944d3f54f45cc9a84cb0482c6a838 : SUCCESS in 7m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/502b74cd4ae34078ae80f1ef4e8fc4bc : SUCCESS in 11m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/83deab878ef446738a092e3f571ed5c5 : SUCCESS in 6m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/46dd9ebfd9e74417a9cec666a0a5d64b : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61cd5dec2aed4a7fadd08acb4fe57c13 : SUCCESS in 10m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2c1a1151639c4dcc8e12528de1291485 : SUCCESS in 10m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/205dfc58cd0d417fb90ea6bfaf4a81ae : SUCCESS in 21m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8f19f5974032478084c6e391b7471ca9 : SUCCESS in 13m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/06474d9201254ce7af546a5fa1e47d13 : SUCCESS in 6m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/07cea4a5aab248d186dae2bfc3159470 : SUCCESS in 21m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/84925cba1f844090b9e18bc48f442cea : SUCCESS in 31m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/537752fe53f6435cb4f35d4971e26397 : SUCCESS in 7m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fbf1387bf463495c9e0f08728c29a5bd : SUCCESS in 6m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3db0ada393c44b0987f7cdf9cc3cae79 : SUCCESS in 20m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7d5a1383399148fbb6b92d10a6f1fa63 : SUCCESS in 8m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6e3b67e0ab2a47dfaf24bff8e97ceec0 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d91ae693359e419ebbd9685430766aaf : SUCCESS in 8m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a817de70b8484648a3539f860d85d9a6 : SUCCESS in 19m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d6caad63b3e48a2904c40d1bfec0fc3 : POST_FAILURE in 31m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af3e5ca146fd4afc90b49847ad83bd4e : FAILURE in 25m 17s","accounts_in_message":[],"_revision_number":4},{"id":"7c794398cfc5b16dd6a334a3a66aa51dc6f69242","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-29 04:41:26.000000000","message":"Patch Set 4:\n\nrecheck.","accounts_in_message":[],"_revision_number":4},{"id":"5019046543f2919ca44d771f54f183f2a35ab7a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 04:57:13.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c45d51f56ca94058a5f1f896e7735e3c : FAILURE in 8m 19s","accounts_in_message":[],"_revision_number":4},{"id":"645f5425b92a861b6b6662460da1e08e97e184ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 05:54:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2e46e4c61742009bfc115bf8cd3a5c : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/249ba55c71004cd09c5a05ecfa4ad91b : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35a4a14705e041089165e9311e10425c : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de910805ea9e4ba4af1a8d9e35355d21 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a72524297f54e14b89e6b4913759b18 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2e6d3d842204a81a1dc841ede69f01a : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b71ce4628b78489a87684590382d99c9 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d722122e31d0468aa4d2bba4d36015b2 : FAILURE in 6m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ab093da958514a6aab5535a029e9324a : FAILURE in 8m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dff764a6627641b29be8802b91937394 : SUCCESS in 6m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9ae0ab5439574608ab6d112cfff47f43 : SUCCESS in 20m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0fdd074a492d458fad898cdb1d8488fa : SUCCESS in 7m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/824bddeee6504111a63884012a427587 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2f6bca38fd943f389fa7220016c4b26 : SUCCESS in 8m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2430f7388a9c4ba29c59ce354a9db176 : SUCCESS in 10m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6dfcacd40b5c4af28063f4382dc7ef45 : SUCCESS in 14m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/24f7f442b29545bbbc5199acbc50cd9a : SUCCESS in 11m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d5a78d31325448ac8a28976df9777853 : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1261ae21a259495c9150de4e66e2d2dc : SUCCESS in 24m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a72853683e9e4dc5adf4c8c123dd5397 : SUCCESS in 13m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c6a25ba7fea842dc8a6645790b3a1bf8 : SUCCESS in 8m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02c44bd5ddf34dd092e7fe9f8b984c11 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a8afdfffeff4574953b7de446f1ee9d : SUCCESS in 21m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5679ef48e2fc4d3f8f8c709a678a235d : SUCCESS in 8m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/80f4d4b06e8e4625a6546bdf2bd2bbdd : SUCCESS in 8m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f8c0c091a03498599ab6f8e458c0250 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f2be13ce9c5340d1b40e26496dfdf86c : SUCCESS in 14m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/580e9cad05c242ebafd3fb408824e67a : FAILURE in 39m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0ded4f3c9104ef48de46310111c2a9b : SUCCESS in 1h 12m 27s","accounts_in_message":[],"_revision_number":4},{"id":"639d46295f153a8070f3a8b979833d74f82490b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-29 16:18:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5286ea94284b57c89c547d0fb329e94acb0dd22c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 16:34:47.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/25a40cd47fb64b0ebc4d34f93dbe45ed : FAILURE in 13m 57s","accounts_in_message":[],"_revision_number":5},{"id":"778d02d6b94ef40c1a6ac25ba923aad47b584beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-29 16:51:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a984916a8cbed62e32a4d0f603f368bdc0dd66e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-09-29 16:52:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"622f4d39d31258fde3fb97067886fc43d95ebf79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 18:25:29.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0677e0fd7b9842d39730b516a65d1909 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/126da582fcd84c7db145267480207158 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89748a37d482424db44e19528da88d52 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1412053b49404c6799da53685aaf5094 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7fbd38e3bbb4678886256425ccde2ed : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1fd46ee6b95431ebe16f1646b8bfcc2 : SUCCESS in 6m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/66f36c0f3c7c45f5b99945a738032fee : FAILURE in 46m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8558340434a747bb9e5bb14fcd038e82 : SUCCESS in 1h 27m 44s","accounts_in_message":[],"_revision_number":7},{"id":"8fc6c72c80a0b93df597c2d6e4518f04cffc5d5f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-10-11 12:34:38.000000000","message":"Abandoned\n\nduplicate of https://review.opendev.org/c/openstack/requirements/+/813405","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ec88b7ebe6db1e05507d07cc9750c1d9753a66b5","revisions":{"389b1a9221b5afaaa6b3c7817d0b12807d581b23":{"kind":"REWORK","_number":1,"created":"2021-09-15 08:04:35.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/809153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/809153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/809153/1"}}},"commit":{"parents":[{"commit":"7d440eb91773aa8a9f455a3f7c69db780e926679","subject":"update constraint for python-cyborgclient to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d440eb91773aa8a9f455a3f7c69db780e926679"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"subject":"[DNM] Test Django3.2 with horizon","message":"[DNM] Test Django3.2 with horizon\n\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/389b1a9221b5afaaa6b3c7817d0b12807d581b23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/389b1a9221b5afaaa6b3c7817d0b12807d581b23"}]},"branch":"refs/heads/master"},"32007c3c8d2f781cce8c9ed0c9b8e2d85f0ca740":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-15 09:37:45.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/809153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/809153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/809153/2"}}},"commit":{"parents":[{"commit":"7d440eb91773aa8a9f455a3f7c69db780e926679","subject":"update constraint for python-cyborgclient to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d440eb91773aa8a9f455a3f7c69db780e926679"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 09:37:32.000000000","tz":330},"subject":"[DNM] Test Django3.2 with horizon","message":"[DNM] Test Django3.2 with horizon\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/809163\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32007c3c8d2f781cce8c9ed0c9b8e2d85f0ca740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32007c3c8d2f781cce8c9ed0c9b8e2d85f0ca740"}]},"branch":"refs/heads/master"},"8e951d679654b0e7d0df94451e59b3fc00f164e9":{"kind":"REWORK","_number":3,"created":"2021-09-24 09:49:00.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/809153/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/809153/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/809153/3"}}},"commit":{"parents":[{"commit":"7d440eb91773aa8a9f455a3f7c69db780e926679","subject":"update constraint for python-cyborgclient to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d440eb91773aa8a9f455a3f7c69db780e926679"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-24 09:48:49.000000000","tz":330},"subject":"[DNM] Test Django3.2 with horizon","message":"[DNM] Test Django3.2 with horizon\n\nDepends-On: https://review.opendev.org/c/openstack/horizon/+/809163\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e951d679654b0e7d0df94451e59b3fc00f164e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e951d679654b0e7d0df94451e59b3fc00f164e9"}]},"branch":"refs/heads/master"},"e0e9069861dbaeb23d406e527050197588880d5d":{"kind":"REWORK","_number":4,"created":"2021-09-28 13:08:49.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/809153/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/809153/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/809153/4"}}},"commit":{"parents":[{"commit":"7d440eb91773aa8a9f455a3f7c69db780e926679","subject":"update constraint for python-cyborgclient to new release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d440eb91773aa8a9f455a3f7c69db780e926679"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-28 13:08:33.000000000","tz":330},"subject":"[DNM] Test Django3.2 with horizon","message":"[DNM] Test Django3.2 with horizon\n\nDepends-On: https://review.opendev.org/c/openstack/horizon/+/809163\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0e9069861dbaeb23d406e527050197588880d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0e9069861dbaeb23d406e527050197588880d5d"}]},"branch":"refs/heads/master"},"0bafef328428c460a5db7399d647313e7f631425":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2021-09-29 16:18:13.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/809153/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/809153/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/809153/5"}}},"commit":{"parents":[{"commit":"a0d5baa46432649d799893ed7300fe78bd3a5e09","subject":"Merge \"update constraint for virtualbmc to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d5baa46432649d799893ed7300fe78bd3a5e09"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-29 16:16:57.000000000","tz":330},"subject":"[DNM] Test Django3.2 with horizon","message":"[DNM] Test Django3.2 with horizon\n\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0bafef328428c460a5db7399d647313e7f631425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0bafef328428c460a5db7399d647313e7f631425"}]},"branch":"refs/heads/master"},"314efbcd5770f276e0837f7c56eff09492940ee4":{"kind":"REWORK","_number":6,"created":"2021-09-29 16:51:42.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/809153/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/809153/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/809153/6"}}},"commit":{"parents":[{"commit":"a0d5baa46432649d799893ed7300fe78bd3a5e09","subject":"Merge \"update constraint for virtualbmc to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d5baa46432649d799893ed7300fe78bd3a5e09"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-29 16:51:29.000000000","tz":330},"subject":"[DNM] Test Django3.2 with horizon","message":"[DNM] Test Django3.2 with horizon\n\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/314efbcd5770f276e0837f7c56eff09492940ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/314efbcd5770f276e0837f7c56eff09492940ee4"}]},"branch":"refs/heads/master"},"ec88b7ebe6db1e05507d07cc9750c1d9753a66b5":{"kind":"REWORK","_number":7,"created":"2021-09-29 16:52:51.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/53/809153/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/809153/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/809153/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/809153/7"}}},"commit":{"parents":[{"commit":"a0d5baa46432649d799893ed7300fe78bd3a5e09","subject":"Merge \"update constraint for virtualbmc to new release 2.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0d5baa46432649d799893ed7300fe78bd3a5e09"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-15 08:03:52.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-09-29 16:52:38.000000000","tz":330},"subject":"[DNM] Test Django3.2 with horizon","message":"[DNM] Test Django3.2 with horizon\n\nChange-Id: I7c22d7cc8092222a91a1da2bd65958bb3a9599eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec88b7ebe6db1e05507d07cc9750c1d9753a66b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec88b7ebe6db1e05507d07cc9750c1d9753a66b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
