)]}'
{"id":"openstack%2Frequirements~809678","triplet_id":"openstack%2Frequirements~master~I48f6af30e7be5a96a79a04182af7cf739a8924c0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I48f6af30e7be5a96a79a04182af7cf739a8924c0","subject":"update constraint for python-masakariclient to new release 7.1.0","status":"MERGED","created":"2021-09-17 13:29:32.000000000","updated":"2021-09-17 16:39:00.000000000","submitted":"2021-09-17 16:34:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809678-new-release","meta_rev_id":"9feecb8f847a595ef8ec89557b9357530ab52072","_number":809678,"virtual_id_number":809678,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 16:34:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-17 15:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-17 13:48:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-17 15:07:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 13:48:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-17 14:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 15:05:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 15:07:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d996aaa45d805fcfdf41835e2295778869d2cbe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-17 13:29:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"391c3b3ecde8507c3bc3cda59d00158fc7cc3112","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-17 13:48:06.000000000","message":"Patch Set 1: Code-Review+2\n\nit have been forgotten during the last batch of patches for clientlib","accounts_in_message":[],"_revision_number":1},{"id":"20d154350f5ca1b3a1adcbc7ae391d104a492840","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:00:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5806cf6201a148dea77ab835414670d8 : FAILURE in 15m 46s","accounts_in_message":[],"_revision_number":1},{"id":"dd4b1b4432e2831fe93605574885d87b5f5018de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 15:05:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e4155b5e6247f690e7feec9f057af3 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7128da45ddf4f6195223c2cf567dbea : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d834ab7f55af4032b2b27fe6a4ac864b : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dce110a7c2e248c59cc05ae3d52e0eb9 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7a475297ff7486fb9fab89c1cbccbb3 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cf9d1766900414fbedd05f69a67fdbe : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3233de4427b94d10a309c6ca15f2d084 : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/031458814cc64ccb9beabd2340cd6481 : SUCCESS in 13m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f4e69e5c3ec04c4bbba8c870292496dc : SUCCESS in 9m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4dc77407db5446a987e98e04115a6a1e : SUCCESS in 8m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/caea42b5390644fa81360c26a4d90280 : SUCCESS in 10m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e1e26afe433f4cffb14adf22db0e2125 : SUCCESS in 6m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/807297ce98664af28333149b32cc0bb7 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff7bfd2f35ee4a118c189f96223fea48 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/85ea268980b446be96b3ed707bec448f : SUCCESS in 9m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b48d4adac3d74387adb1d494a0c8e786 : SUCCESS in 15m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/21557dc195034323ba96a5f310cca846 : SUCCESS in 12m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2fc1927bfdf94bc8baf6509162b96135 : SUCCESS in 6m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e6d32aef60d84ad4aa9eda4fadc62e79 : SUCCESS in 22m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c25ae3973ae04e9cb7d2350fc85d4060 : SUCCESS in 11m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c88a2573db25417a83c37cf96a421254 : SUCCESS in 8m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/15bb4ac48bc940dc95887b9e7648efb2 : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4c26e3557b494dd0a025477f52f898cb : SUCCESS in 17m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e66f04d71aa645a2b4bbb781d67fffb7 : SUCCESS in 7m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6251e6eab5db4df89798f2359e0db363 : SUCCESS in 8m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c860798e310b46e9b6641d41804fccb2 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9f0b99eae93c45e58b7aef2385fc49d4 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81db655b8fa94449bf36ef0e4c66ee01 : SUCCESS in 48m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0a14841ae6c46d299ebea7867a97dd8 : SUCCESS in 1h 21m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1e9f3e20ec8cda895d4a11cd19260f84667cd7e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-17 15:07:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d4da8b534fac64e8fa115794dc8cc77f8390d67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 15:11:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"55c481d8bedab1010ae62e75c448701d751fab05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:34:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c38b84f92ee481d83b57114ba73897a : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/641bbed0d20843de8d7d5e573660aad0 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34f0d0ccb81440cfb8edfe2b65b73f90 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb2775f3aca54a3ba6f97b8dea3ac418 : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fceb99fac7474dfd893767d2be986be6 : SUCCESS in 5m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b06e627777c0492788845596f64f4ca6 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a79fd049e21b4ef9a9c8c5d4d28f13d6 : SUCCESS in 13m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/065254e4aab0453eb22f604946cee077 : SUCCESS in 9m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e38e919ad1ee4553859129405d91a6a6 : SUCCESS in 10m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dc793556b25d42db90d8199e5797c12a : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/35996961528046a6bb63305b692d332f : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/502f7e489478418a8c90a5be266ee56c : SUCCESS in 8m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0c1e4a001c6045b69ebc0f6bddf00d49 : SUCCESS in 10m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/99813065250d442aa9f69bde263eb9a4 : SUCCESS in 13m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bcd52c69b11c4eeab15a95566d409423 : SUCCESS in 9m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2e0fa0b30a6a4678879e8b5862fa4fe0 : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/741d67c73cd94ea597c36b74e9a5de08 : SUCCESS in 24m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d228c0940f4542ec8dc5488ed6d21931 : SUCCESS in 11m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1420199d21df4e16b44597a154d967ac : SUCCESS in 6m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/757e61fed2a448bab13ad6b7f33563aa : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4734e35900a64a96a9429439971e8091 : SUCCESS in 17m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/bbece313324a4f5aa2e03a0e557e995c : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b3015de2a6a743d88d9f4fe9cab5e094 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3042e99a4d4e4b3097fbcd261defa9f8 : SUCCESS in 9m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/87ab84791cc8453091116a55137358ee : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71ebbe2e7f9b4f63805510a728350de4 : SUCCESS in 52m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f1c749b63a24383a9ef7c2751b4596f : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":1},{"id":"14733845fffe17bcb43b94620b9cfd8014def02e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:34:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9feecb8f847a595ef8ec89557b9357530ab52072","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 16:39:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d4171b509b473686d7b55b2a663c55 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8825c3fe73f084fafe3e339ec60ddadf82bc656","revisions":{"c8825c3fe73f084fafe3e339ec60ddadf82bc656":{"kind":"REWORK","_number":1,"created":"2021-09-17 13:29:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/809678/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/809678/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/809678/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/809678/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/809678/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/809678/1"}}},"commit":{"parents":[{"commit":"863eea3390630f9e226367d4ade1c2ebd5c2a216","subject":"update constraint for ceilometer to new release 17.0.0.0rc1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/863eea3390630f9e226367d4ade1c2ebd5c2a216"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-17 13:29:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-17 13:29:31.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 7.1.0","message":"update constraint for python-masakariclient to new release 7.1.0\n\nmeta: version: 7.1.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I135919402c7d867ad0ab8b800d7913f7f2113d35\nmeta: release:Code-Review+1: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I48f6af30e7be5a96a79a04182af7cf739a8924c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8825c3fe73f084fafe3e339ec60ddadf82bc656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8825c3fe73f084fafe3e339ec60ddadf82bc656"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
