)]}'
{"id":"openstack%2Frequirements~809879","triplet_id":"openstack%2Frequirements~master~Id4fd5f8c70e3042ea9826e9999e3a33173cde628","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id4fd5f8c70e3042ea9826e9999e3a33173cde628","subject":"update constraint for python-freezerclient to new release 4.3.0","status":"MERGED","created":"2021-09-17 17:13:49.000000000","updated":"2021-09-17 20:26:46.000000000","submitted":"2021-09-17 20:24:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809879-new-release","meta_rev_id":"53fe0a258f66603225e104755fb2f929792b0007","_number":809879,"virtual_id_number":809879,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 20:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-17 17:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-17 17:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-17 17:39:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 17:16:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-17 17:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 17:39:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-09-17 18:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ad289ae5b916d233e31bea831f2754261e41d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-17 17:13:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ea72513bb14381f77a591742a11f4ced34c2cd7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-17 17:16:41.000000000","message":"Patch Set 1: Code-Review+2\n\nit have been forgotten during the last batch of patches for clientlib","accounts_in_message":[],"_revision_number":1},{"id":"d08086ccd83eccebd238ebd9a12e72f18ab9d6db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:33:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8b638680a02346bd902c79ff0181bb06 : FAILURE in 17m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f0fbb042ecd3685363ca500c48dc89579ee8e186","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-17 17:39:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9292b44fd1b42011a9fb0eb621e0184e2640fd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:55:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da8c97381c54fe89d83785cc3974a56 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dc290b55ded411fb8f2195a24db0d8f : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5e4f12758ec4f5e928bf8c98f4f5a57 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04ebaf5e60164736b274f886689f30a4 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3bcd22e7c7f4a5f8dc851dd9bc317d2 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/444257ee68634b2a9384c7393dfc5a4b : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9abfde90d7594df98d557c4170f272de : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d25f2adc0099461f97f725c57a8f9854 : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bf8e67c083354ffe84a74876370010e3 : SUCCESS in 9m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6327df63f6eb4eb08d867f4448e14f05 : SUCCESS in 7m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b937cd3e99a84a55a72944db5a198b09 : SUCCESS in 12m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/234828335a2442d7a67b283b6e9b7592 : SUCCESS in 6m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ff87ea0d41f54806bdcdd4d6e8d37c35 : SUCCESS in 12m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27031c5ac98c41b2bf06d4cf1dcfd5af : SUCCESS in 10m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f27288901d6743949d69fc3795d37b98 : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/385a884a37934b64b3224d45be30b86f : SUCCESS in 14m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c0f51c17f593461d8c91ede58ea01271 : SUCCESS in 11m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ad87c1c09ec449e99c5ea2f33b06baf5 : SUCCESS in 6m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ade26111036d4daab37a0b5c554659b8 : SUCCESS in 20m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/256a4d03f5e84df194057e7dfe8564bb : SUCCESS in 13m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f8b7ef87b84246de823c517123ada356 : SUCCESS in 7m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30f72952f37e45409a9730991dab8963 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/938872560f944098b473abc3fb19d3bd : SUCCESS in 21m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3d541cf20c3c4c26803b037f85142c3a : SUCCESS in 7m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b2e4fc2370154328a5e7d08e538cfaeb : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6e32d0c54e048ef9f25cf3912f96218 : SUCCESS in 8m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8a727f71daef46859078be16d063f4d5 : SUCCESS in 15m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67c8fb7228b944979f3227fea5cab6a3 : SUCCESS in 50m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5e3da2264ec49619eaa5ea9eeeb7420 : SUCCESS in 1h 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"cbb26933a26a4199a4b630d157aca7a728d746a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:56:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f60548d3d6c040eaa5891e58b2f47f20aca78ece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 20:24:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd4dd487dac43aea32c51a3a227f6c6 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac42fd85ce8469d8cfea25a7f216f6e : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/549a2ed6299b4941ad94d6191c073674 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9c5e7a5522442138e1cec0b16fb9012 : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad8ecb28c2ad45bc93ab6d0ea9a6a89a : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94fb86004bef476cb206ea43363205fe : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2fee3733a98b4253b36c8adbde397e75 : SUCCESS in 12m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e43522a40dbe47059f6c11eb63a4f8f6 : SUCCESS in 16m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/de3ab5c56b6444969895968551120966 : SUCCESS in 11m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/46a37f909e214a58885e719334c88463 : SUCCESS in 8m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4f2b8473c7fb45589ee129c1e7e37e6a : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a173b4ebf2994f7c95a5723c5d19b944 : SUCCESS in 30m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/73450e1ac4fa486abdce57f044b146f8 : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5352e65655644eda8b7de59a97046d52 : SUCCESS in 14m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d2adf1200e7c4060809710b647088dd5 : SUCCESS in 11m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a8120ecf61ea4ab6b001f9dc81ffa04e : SUCCESS in 6m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cf107c747f9949cba0d7810423996c27 : SUCCESS in 24m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/80897e15b43d4cce97ed5c65e74446f8 : SUCCESS in 12m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cc3afae5721e412f8e59be6b2bb40e6a : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f9dbabdbb7049da9c8599e158e66df9 : SUCCESS in 5m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df06d90342a34e89ae0bbb727632c1d1 : SUCCESS in 20m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/29b091cbf42548aa9e5bf349e8abf4e6 : SUCCESS in 8m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e3fc5d6cea4d4b6db050df038e3659c6 : SUCCESS in 7m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/78643a3637e9448fabb5851c2e581fe2 : SUCCESS in 7m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2d8ab43e509c4e7c90229ae378243016 : SUCCESS in 13m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e59748a715094a6db755b0faaf3d154c : SUCCESS in 44m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ac17ad204424d78a778d2e85ffec2ab : SUCCESS in 1h 22m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5ed10c6111bcb1f9d8a6ccd77081aac1187d9f4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 20:24:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"53fe0a258f66603225e104755fb2f929792b0007","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 20:26:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f24ccdcae234ee8ac6227de82e37a87 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"013d14332b92ebb0d7a34541e96093400fb049c1","revisions":{"013d14332b92ebb0d7a34541e96093400fb049c1":{"kind":"REWORK","_number":1,"created":"2021-09-17 17:13:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/809879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/809879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/809879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/809879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/809879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/809879/1"}}},"commit":{"parents":[{"commit":"c8825c3fe73f084fafe3e339ec60ddadf82bc656","subject":"update constraint for python-masakariclient to new release 7.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8825c3fe73f084fafe3e339ec60ddadf82bc656"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-17 17:13:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-17 17:13:48.000000000","tz":0},"subject":"update constraint for python-freezerclient to new release 4.3.0","message":"update constraint for python-freezerclient to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I04733290d003fef0fbdc227e3c7f16899acb3013\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Id4fd5f8c70e3042ea9826e9999e3a33173cde628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/013d14332b92ebb0d7a34541e96093400fb049c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/013d14332b92ebb0d7a34541e96093400fb049c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
