)]}'
{"id":"openstack%2Frequirements~809882","triplet_id":"openstack%2Frequirements~master~I2c0832fe7abe1ebbc88a8bc3fa1fbece1e8d76bc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2c0832fe7abe1ebbc88a8bc3fa1fbece1e8d76bc","subject":"update constraint for python-barbicanclient to new release 5.2.0","status":"MERGED","created":"2021-09-17 17:20:54.000000000","updated":"2021-09-17 20:24:12.000000000","submitted":"2021-09-17 20:24:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809882-new-release","meta_rev_id":"5e5f3e87ac81664085f079b329129819498f1585","_number":809882,"virtual_id_number":809882,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 20:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-09-17 17:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-09-17 17:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-09-17 17:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 17:22:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-09-17 17:35:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 17:39:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-09-17 18:45:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84598315b0291444a41d61a6a35d0c72f345caba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-17 17:20:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6a1183371462450857b744d5cc770922f303430","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-09-17 17:22:41.000000000","message":"Patch Set 1: Code-Review+2\n\nit have been forgotten during the last batch of patches for clientlib","accounts_in_message":[],"_revision_number":1},{"id":"b53458abb829571e93832dce067e355065b2c475","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:35:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/836b10206f2c4b7eb2700f69b241aeb3 : FAILURE in 10m 39s","accounts_in_message":[],"_revision_number":1},{"id":"be7c762d20445a57991e54b97a8c80abc167abdb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-17 17:39:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65710cc6e55f6226f72f6278f7812749e46919e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:45:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb81d4081524046b4d48c26628cdba7 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9ff5ff7e8f447e9825e5682cbdf11c : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e396a45656e46fd8719fc25ee03d754 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0da8a73a1fc14d3fa63685c45bef0a34 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9e2347e7e9a477992d3471c9ab8796e : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc1cfdd7ee204308b0b19ac753ed800c : SUCCESS in 6m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1730a2e31adc413bae78ccc9e0c82ce5 : SUCCESS in 5m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93ccd12230e74700b44f7b54c93691a2 : SUCCESS in 12m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c1015b19e4b94cfcbbab9ff4396d3f9e : SUCCESS in 9m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/70872c59d43446c2a4742df7f120e26e : SUCCESS in 8m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a2d27b12a0cb43a692cfb286d5aa9c1a : SUCCESS in 10m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e8aad40d1a5e4829ad4a3c8ddfb6f6c8 : SUCCESS in 6m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ae7c27cbbbd74fc4800e07e9b8c42bfb : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dd5d76a5dc574efd9f369593c5b15e85 : SUCCESS in 11m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/616a1772fecd4c248d4ddbbf21efa42e : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ce20f5476f8d413698f7fcce8f31bb33 : SUCCESS in 14m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5a9d1a61c6054311bd7858c6eee5aa94 : SUCCESS in 11m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3e838c93ded5490fb52e1fbf9256e005 : SUCCESS in 7m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f3ca912d489449fc948ca3b738a3ec9a : SUCCESS in 24m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/59d72670548748178ce88d7dad7108ba : SUCCESS in 12m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/50b2c41287ba4ab883ce1fdb7043114a : SUCCESS in 7m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4631e33e2a854c9fb331a78f85eb6ff4 : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/36d2e595af8443bb8de9319a518ce412 : SUCCESS in 17m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4ce9b318245f486c8035c13a6a8dc5b4 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/254ef81cf1f141bd8c1dcf196ad5813c : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55f127240340439d92b3cf106f647aa7 : SUCCESS in 8m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f1f549cbad804858a540b901ced24a68 : SUCCESS in 14m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed6ff4499319496ab111e61f2a531a02 : SUCCESS in 39m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/020ca12036ce4838aded87bd62f662ee : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"feb7bbe98fe0a38c38fd8334f4ee54d023fe972a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:47:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b11a6157565183cbb20fc1d4a20ee1d0f50df2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 20:24:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ef6a96c2fb418891bfd5ac596430a5 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642bd15d110e49d688662c56cac6f0b7 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4b618073959420a97626ab0ccecd7a2 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2cbf23c1a294b80b8f611c527008b71 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/06918d3aeec24a65a2da353e67cda071 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5daf4201d23c4fe78adbf6e41daa6c9a : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6872607e1764e42a1c4bcf1d9559c91 : SUCCESS in 10m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e2e005e0e79411d8c02f41172713914 : SUCCESS in 10m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/97ff29de61f141008d3bff9d2bad52ab : SUCCESS in 10m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/385f0b720e2e4d228be3d6d85f627f3c : SUCCESS in 7m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/90484fc750eb43a386410668337ee9e6 : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c5ed159530ce452dbd00ddc23f17ae9c : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/53701043ae904a82a3523eb17f51645e : SUCCESS in 9m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fb2d4a0f313a4f7b96d8dc39614b9d2d : SUCCESS in 14m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5eaf7d78ea9c4974b03971015cded883 : SUCCESS in 12m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/92bee2e2bb7c4ecfbac3b5dfbea7e686 : SUCCESS in 5m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4300577331e14119b6252a4978d4d5d3 : SUCCESS in 26m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20630a8a85e8484b88235e8ad28ce047 : SUCCESS in 12m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b3bfb334350345a49a9bffb4b364d734 : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/05ae55ba7f934cb598d4f91df6213a5b : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f718e990253f4e559458a31b1301ac9d : SUCCESS in 18m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/50dde70bb5924c2f83c6876ecd0344b4 : SUCCESS in 6m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/298230e35c4944c1a17cf6e6229ffe79 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a60e83161bf4d2e84a0d82d19ff5004 : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b15c698d75ff453b9cd5fd80ec2ea90f : SUCCESS in 12m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f1f3cbc928a4bcbbf98a6617bf6acbd : SUCCESS in 50m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db0ba9c74de74f4a906b99552e5342f9 : SUCCESS in 1h 26m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5e5f3e87ac81664085f079b329129819498f1585","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 20:24:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9291ae82b4d737b6eb25fb30620de94b2fc79184","revisions":{"9291ae82b4d737b6eb25fb30620de94b2fc79184":{"kind":"REWORK","_number":1,"created":"2021-09-17 17:20:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/809882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/809882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/809882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/809882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/809882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/809882/1"}}},"commit":{"parents":[{"commit":"c8825c3fe73f084fafe3e339ec60ddadf82bc656","subject":"update constraint for python-masakariclient to new release 7.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c8825c3fe73f084fafe3e339ec60ddadf82bc656"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-17 17:20:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-17 17:20:53.000000000","tz":0},"subject":"update constraint for python-barbicanclient to new release 5.2.0","message":"update constraint for python-barbicanclient to new release 5.2.0\n\nmeta: version: 5.2.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I7ba606e6b3a0d3c24982e7bee442518c4d4db5fc\nmeta: release:Code-Review+1: Douglas Mendizábal \u003cdmendiza@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2c0832fe7abe1ebbc88a8bc3fa1fbece1e8d76bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9291ae82b4d737b6eb25fb30620de94b2fc79184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9291ae82b4d737b6eb25fb30620de94b2fc79184"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
