)]}'
{"id":"openstack%2Frequirements~809971","triplet_id":"openstack%2Frequirements~master~Ic8841d9bd31a9aabd5232ccabc73dc2348a5bd0c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ic8841d9bd31a9aabd5232ccabc73dc2348a5bd0c","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-20 06:54:05.000000000","updated":"2021-09-20 15:32:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc693b82a0997f5e833e09fcdf6092a4d6792a2b","_number":809971,"virtual_id_number":809971,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-20 09:10:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 07:12:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 09:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"049851833ec4ad8e183ed9c06fbf367e000cb6ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-20 06:54:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da17eda27e09f93b786ae9544e5e656e7531e61f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 07:12:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/77b55e6d265f4220ac92fd5f566d89ae : FAILURE in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3a7449b904cd59d62c10116eddfc4eab531f48bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 09:10:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9a3e4b179e4895a629fc8a24900bb6 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/824b25cc06534a5d8a821b393002460b : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3d766728c744a12addaa169aa665b9a : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e21ccc459f8c4a4d9ceffaaf13d4fc62 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d45c03576e0a492899d7925bfe5e63b4 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb80627157f54097be1a9f110db02ce2 : FAILURE in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0763f10fd9774775831c2c11e4e5561c : SUCCESS in 5m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fdeaf20ef2524785a371f4e14b6aaa8b : FAILURE in 5m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8fabf104765749e48714614b9d5ae790 : FAILURE in 5m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7b4aceab2ae04917bb9bb9d287480406 : SUCCESS in 8m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/07778c870df943c9b09debb6624fc769 : SUCCESS in 11m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/696d7aa84db5414a9c1b25c2ad18b888 : SUCCESS in 8m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5a78cf66560a45ef83ce03d9a3a885ec : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e9244e7e94e44219b1517c740a88d617 : FAILURE in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/112b05f7b8b3438faf8975dc8f2b14f5 : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/174f6363c8f846a78097b68d1f4c32d1 : SUCCESS in 22m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cb873d30123849d9987a81acd2f38895 : SUCCESS in 19m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0954c0886afe4ec48b1a716ce406633c : SUCCESS in 8m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b2a42b0d3d8646eea0c9267c7f4877ae : SUCCESS in 23m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7ad0eab855c14323963bd6d9c75e5aa5 : FAILURE in 13m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2868a53fad7a4670abd5aa04db48fbdc : SUCCESS in 7m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/786ea871fd5e4f0f83fdc7a3471a28a3 : SUCCESS in 5m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56679876162443bd87e9d604eb2e7bdc : SUCCESS in 20m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5acd7d7d4c844f1abc10428a0b602f2c : SUCCESS in 9m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b39a9af4684a4339b50b8bbabdda9dfa : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5ff61b45f444b42a09bb2c544bc5b47 : SUCCESS in 8m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ca7206f08eae4d8ca5f99de34cb9c654 : FAILURE in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/485eb35b444c4516b5ebf9227460aa2a : SUCCESS in 45m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8d423634b8848dc869afded8e0e30a6 : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":1},{"id":"fc693b82a0997f5e833e09fcdf6092a4d6792a2b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-20 15:32:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"afee452eff1bac926f8f249a6130f805f7285720","revisions":{"afee452eff1bac926f8f249a6130f805f7285720":{"kind":"REWORK","_number":1,"created":"2021-09-20 06:54:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/809971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/809971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/809971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/809971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/809971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/809971/1"}}},"commit":{"parents":[{"commit":"e93daf2c3e2fc139d85a679999f846196c8ef2c7","subject":"Merge \"update constraint for keystoneauth1 to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93daf2c3e2fc139d85a679999f846196c8ef2c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-20 06:54:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-20 06:54:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic8841d9bd31a9aabd5232ccabc73dc2348a5bd0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afee452eff1bac926f8f249a6130f805f7285720"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afee452eff1bac926f8f249a6130f805f7285720"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
