)]}'
{"id":"openstack%2Frequirements~809975","triplet_id":"openstack%2Frequirements~master~I80ef7d9a20e007b6db965b37113870ce8a5a68c4","project":"openstack/requirements","branch":"master","topic":"prettytable","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2021-10-12 08:19:42.000000000","reason":"stephenfin replied on the change"}},"hashtags":[],"change_id":"I80ef7d9a20e007b6db965b37113870ce8a5a68c4","subject":"Remove python-ceilometerclient","status":"MERGED","created":"2021-09-20 08:52:54.000000000","updated":"2021-10-12 11:17:14.000000000","submitted":"2021-10-12 11:15:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809975-prettytable","meta_rev_id":"1c8eaac5027fb1ce62c6eb06323cbe8d5631a6c7","_number":809975,"virtual_id_number":809975,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-12 11:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-10-12 02:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-10-12 02:35:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 09:13:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-20 10:10:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-20 15:33:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-10-12 02:35:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a15e8fc414107cb5afb73cd38df23b8405d82f78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-09-20 08:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e09f8f43143358c0db6252ba34fad96a8437ce22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 09:13:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dbe72e89aeb04dc9afc88eb67d5eb28b : FAILURE in 18m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7c4fd6a681ff64da4df4e2ee109bd84da331efbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 10:10:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8563bc1c4a274306b7b33a072a3f0be6 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a3e0861e9544adbf757ec80af4480a : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e500ab91a8a461a91a3a1d451a3f7f3 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/433c6e7a440b4bc4af74d90b1743d6aa : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac17ec5c11a24b5b9b13fd873179464a : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb6e0d524bab48d9a8f1150f367d5b5a : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32131912c5a64ad3a7f31c9c6805050e : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0de70e53e1ea414886b287343b407f2e : SUCCESS in 11m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a181f60e96e0490ebdbef46e3bb41d4b : SUCCESS in 11m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e36f1a4b7dfd4fc585dc9b4a68e2fe38 : SUCCESS in 6m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/73160469899b40f0957ff205e2947137 : SUCCESS in 11m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2c437978da154f0d985a54807d8ff5d2 : SUCCESS in 5m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/03e93bc616ba46ca8ba4b1716903e6ea : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d738fbca69342db948d6639acd395ea : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/481c3cf644ce48889a781460809bdd43 : SUCCESS in 8m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e8201ad89aed4e75a11857bdff350181 : SUCCESS in 18m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9a9b4c909e574994b43fc51422bd7700 : SUCCESS in 10m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b0ee6b834b764815ad3d70d256e859f7 : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/45fa6c1993844c7985856ab931f12f6c : SUCCESS in 29m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3f39f6335e9c4468bfc5912c0e680195 : SUCCESS in 12m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6a56e531ee5945f98e33c484687fa09c : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d714e7024f3f4bf5b32224c56fbef82f : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3b969e90576344369e52e09307b9bbc1 : SUCCESS in 18m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e9975c6b4ac54f8a98d66c5d0599c4d5 : SUCCESS in 9m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e67e4b0b8c0f408c927e0f1c76ee628d : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b59269ffb8cc45a4ba39b498a18bfcff : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0cd52873b388414eb748be694ff558ea : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/908d3b3cc9a24990bcdcc58859e4b878 : SUCCESS in 41m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1db4fe0278540f796244096cd6daee6 : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0cacfe5e4ee4aa08954c65ea5ff8938906ee7366","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-20 15:33:28.000000000","message":"Patch Set 1:\n\nyes, but...\n\n| openstack/heat                   | requirements.txt                                         |   39 | python-ceilometerclient\u003e\u003d2.5.0                                                   |\n\namong others","accounts_in_message":[],"_revision_number":1},{"id":"43163b0b49393ab1e429ca0f9f4d0293ef02decf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-10-12 02:35:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ec32d4e92811b54346847fd80df8ce9952e246b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 02:35:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc3669d4785232284fc3f396ee2f38cae96e33f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 04:00:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5283abf0a9a3402f91fde4ace2d3fd52 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e918709496545c5a8c8fe50922d7ce2 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e105bbc744e545f7bfeaac430b5c89de : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f715409950e41bd8c8e93cecb32535d : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0d92bb83a5e4a51bb367dee060c11b2 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14757b440d784262a7518cef2c40fbb2 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d30ce67bb69a43dab01070ac5c41f81b : SUCCESS in 12m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ad66e92611646c5b1d52db77b7151be : SUCCESS in 12m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/88df928c20ae4e868dc50fded59fc41c : SUCCESS in 11m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bd445c60951841618cc484984f3b2716 : SUCCESS in 6m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ef513ab858454b0ea3aea32f7cd06863 : SUCCESS in 12m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c045b7a49ee460c87f9dab96ef59494 : SUCCESS in 9m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ef0f05d84d354f6e8dc3502bfb6ec2bc : SUCCESS in 9m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7acdcd6fdfda48f8990f262b494aef18 : SUCCESS in 17m 09s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3191fa52c39a4d368616df6925a9121c : SUCCESS in 4m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/da33fe5382b844f5ac3ddfeec5515cc4 : SUCCESS in 11m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a8bd54290e9043afaf63300b8fba0035 : SUCCESS in 6m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9e5c38be59ec4356a59476f2b1852c63 : SUCCESS in 27m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eb94c1b8afc0412f9685336f6e596e02 : SUCCESS in 11m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2aad951bc62b4174916fed71874ded7c : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/37fbf0e31aa54610bfdd4ca3eb61c2a7 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2eb92a0eaf94a68bd1aa86b2b3cfd0f : FAILURE in 17m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c6779933a5a343779567c2c75e7573ee : SUCCESS in 6m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0c597cb0f7c14e98bc74158e0b9ad372 : SUCCESS in 13m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/274b3d621c8c45b9bbe13e3026c4e344 : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d0882b9d157e4ba5b30b9ca5d802b40e : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4123c2cd1b7a4f5dbd75e0d91e8a789d : SUCCESS in 47m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f7cd017849f401d824123bf9311aec2 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1adfd22e019370251f5199d017697b3383d9dd63","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-10-12 08:19:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a42dbfabaf340321e1bc59fcc3c8685ad38238aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 08:58:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a69e8f7d462a4493933831122e9de780 : FAILURE in 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e5b1b11dd2a3ac86c81b1666af09b0e2dbb8f5dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 09:44:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff45aeae0f8d45d7987998deaccafbd8 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cdc07b3629947dfbfd90cc48609ae8a : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed8475e0e90e459890aa4b969350e436 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d66b9fd354194262a960ee96a006e3ea : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbbe6da0ace94d03b8c5ec53de3fea74 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1129d211bc7345e09949bfd8ddcc0ae2 : SUCCESS in 6m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9cf4a9a876c64fb4890fad4bd92f7d3e : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/575a6267d53846a29b2cdebe134cb1ce : SUCCESS in 12m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9be509c73c6c4befb2cf6f28272d6b53 : SUCCESS in 11m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/843246f0d88149c5b6dba755f5b1f52a : SUCCESS in 6m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/80fd508c7b644820adc818375fd1bccd : SUCCESS in 11m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d78891aa47bd4c9babe9b751f36f4674 : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a87d3fb33cca401aaa0c33df70e642cc : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/488ed27f422549898a51ef3b407942c6 : SUCCESS in 10m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d15865a60e774eb69a8fdef8dce96f2a : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9bbdadb6535b4ea4996d8d72ea0b4a8b : SUCCESS in 14m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b9f1319fc5ef442aae6b4c74805d7f07 : SUCCESS in 5m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fcae0465fb9d462b9a1a3ad6d27ce9f4 : SUCCESS in 9m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1495a786b4394a8eb8db57e1eddcc8ad : SUCCESS in 7m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4f8ee8d7cc104d88813ac94100cf9426 : SUCCESS in 24m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/48592503e9484c68b273b6d324fad4fb : SUCCESS in 12m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f2999a90a9e64d118ab069936831dda9 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f703e150adfd45078542021dc062b855 : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7751c9da14514ea290af461fc6f69035 : SUCCESS in 17m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ce048b248c0845838033c3340150bc51 : SUCCESS in 6m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/67845640302f4411b490c9fe975f0b1e : SUCCESS in 13m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b91f48071eb044b984d3dc70c0dbd695 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7201f80aa7eb42c58f8ff8c4c6f10816 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90590dd34ea34f07b28e1ed6599f46b6 : SUCCESS in 48m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c76d8ad286d344bbb3e52207c66f4afd : SUCCESS in 1h 22m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e630c70468baea66157dc284a95a883e4d61fd81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 09:45:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3757092f2d234d93c1ba6fc5be4d6db38333b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 11:15:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61413f1c6c0347efbdafe96b16a13201 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f81b3d770742db86e1a3407e3ca48f : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81af2d2ab08e462e96a2cb18a04d2b12 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/599adf3548fd4a56b6e314a7a773b34f : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dbfce99f85964e6fa40cc97939f052a6 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5402addba9574016aac348e75a4029ec : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a793d61ab7704532a25c0af022e27a0c : SUCCESS in 19m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00de98fa85244cfdb55c87186d7cc260 : SUCCESS in 10m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a649fb8c618b4008b1b5e89a8e29f901 : SUCCESS in 10m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ded3dd723a7e4ba08a8637ce7e378d79 : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/059cfec882f54267bf0057137a116eb7 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/593ce58f451f41b5a29df875bbd5c81f : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a501437916a0448d9c2393b1e04eb4f6 : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/05f98fda460047e7af0b34fd2f9abb0b : SUCCESS in 15m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a17afb46652442a18d77b907fb5e16d5 : SUCCESS in 5m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e8f6fd02475b415b8166b501cee94910 : SUCCESS in 11m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/467bc7a8f37045f4a0ada7dd947b6630 : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a96ecd20609e4d25a26797467846c0d3 : SUCCESS in 26m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/333fbab26769493294547d1e353b4bf9 : SUCCESS in 12m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5d33244718334035bcb9e2653631ebef : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca1fbafcb4a94ba6a929709c9dc3ec67 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/edab81bed4f24805843f14c4541d16d5 : SUCCESS in 22m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1f7ebe838fc44415996247623feaa83f : SUCCESS in 6m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f8b6f44ef5ff4ce0bf7c60675219df74 : SUCCESS in 16m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e145e0bb8d4440bac2ca5115ee6ecaa : SUCCESS in 9m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8e3d319eea584db98061deb3aeaa70ca : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bde8a1c9bfb74504bc3b7ba13abb90d2 : SUCCESS in 52m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6189b3f1100945a4831bbd5faeb6965a : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4d48141d224bb89607269f7248276018fc123a58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 11:15:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1c8eaac5027fb1ce62c6eb06323cbe8d5631a6c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 11:17:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/927135d5494a434b940bed88b943669c : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed9671601d162982c2def3ea55fd9796364bef1d","revisions":{"ed9671601d162982c2def3ea55fd9796364bef1d":{"kind":"REWORK","_number":1,"created":"2021-09-20 08:52:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/75/809975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/809975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/809975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/809975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/809975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/809975/1"}}},"commit":{"parents":[{"commit":"e93daf2c3e2fc139d85a679999f846196c8ef2c7","subject":"Merge \"update constraint for keystoneauth1 to new release 4.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e93daf2c3e2fc139d85a679999f846196c8ef2c7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-20 08:51:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-09-20 08:51:19.000000000","tz":60},"subject":"Remove python-ceilometerclient","message":"Remove python-ceilometerclient\n\nThe python-ceilometerclient project was EOLd a number of a cycles ago\nafter no activity for some four years. Remove python-ceilometer from our\nglobal requirements so we don\u0027t test compatibility with it any more.\n\nChange-Id: I80ef7d9a20e007b6db965b37113870ce8a5a68c4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed9671601d162982c2def3ea55fd9796364bef1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed9671601d162982c2def3ea55fd9796364bef1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
