)]}'
{"id":"openstack%2Frequirements~810146","triplet_id":"openstack%2Frequirements~master~I27a3a4fba6a4108ff232676b53ed8fdedc6c1cfe","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I27a3a4fba6a4108ff232676b53ed8fdedc6c1cfe","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-09-21 06:40:08.000000000","updated":"2021-09-21 15:17:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"182067a29adb773de207f3653df32c23ad442051","_number":810146,"virtual_id_number":810146,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-21 10:21:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-21 08:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-21 10:21:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1910edb0aca960a5a22d3c70ef005fec7378bfda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-09-21 06:40:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"487514efcf7b76f1529493949ee850ce9cacf98f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 08:11:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e08600247a94450d873ac564069953fa : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":1},{"id":"00a3b488de473b495b8faa8bcf6f05a5783d7c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 10:21:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0854a1043e64253bf4a2c4babac0083 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8a4dd107be49699f685fc0cb2f6fd4 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54683e252dfb47bfbb3609ae1e8f5ad6 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35dc6307cacf4d51a35222357126d8ac : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77f99dec00f3475fb874f1682f31bc23 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7dccdce7c84491ba2af18cbc432a282 : FAILURE in 6m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/903e159ab9ac44719b868db4947867df : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6d5f26de45c54785a6812f846906f1d7 : FAILURE in 5m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1f85e98e882c43bd9ab68840d4108447 : FAILURE in 5m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/145ff7ea5e6a449ca4c77c8ab4395574 : SUCCESS in 7m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/11162cba9ff943e59b06f314c0aca757 : SUCCESS in 10m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b7324d28fd7946e3bfb7578e2b66b13f : SUCCESS in 7m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7eb71bf79ab24439b68ecaa555f74858 : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/94631d71bcf64bfc9e0a8e834d39fec5 : FAILURE in 9m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4e4c632115204c64b9b516b1f4653854 : SUCCESS in 8m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d575b10669be498498dee0ee6e95902e : SUCCESS in 14m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fffc558c4cc448cd84cd9ac67f71c4ea : SUCCESS in 10m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/68c7466f9526466ba091abd83573a0ed : SUCCESS in 6m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f8a9b36bfe274f63a82add360dfffd98 : SUCCESS in 22m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1cf950abb534449eb4d0750b6a892be6 : FAILURE in 12m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c4c9cb0146ce434fbc518fc19061c7a2 : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/61478684269c4f76837ae8277b1e2915 : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2efba3df15e540fd91afca2f00fc4c6f : FAILURE in 21m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/519e32bf46d345d996a73b42d228a38d : SUCCESS in 8m 27s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8d6ceb2198ea412f908ec24f5dea9b4e : SUCCESS in 7m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55f331d911774253bdc7c8b8091b6685 : SUCCESS in 8m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/083d75d8ca1b4a9d957c9cdb1d4162f6 : FAILURE in 16m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3594b57bea34e939bef358b9b3455d0 : SUCCESS in 48m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eaeaab7eb4c64b7d8d39a9d502e61d9e : SUCCESS in 1h 18m 50s","accounts_in_message":[],"_revision_number":1},{"id":"182067a29adb773de207f3653df32c23ad442051","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-09-21 15:17:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2da7644e0ea3a63fd8b54f418d4b83574e661165","revisions":{"2da7644e0ea3a63fd8b54f418d4b83574e661165":{"kind":"REWORK","_number":1,"created":"2021-09-21 06:40:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/810146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/810146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/810146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/810146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/810146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/810146/1"}}},"commit":{"parents":[{"commit":"4dc5de7a2f50fe9d34516d72dfb455a39766a410","subject":"update constraint for openstackdocstheme to new release 2.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4dc5de7a2f50fe9d34516d72dfb455a39766a410"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-21 06:40:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-09-21 06:40:07.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I27a3a4fba6a4108ff232676b53ed8fdedc6c1cfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2da7644e0ea3a63fd8b54f418d4b83574e661165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2da7644e0ea3a63fd8b54f418d4b83574e661165"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
